)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~688511","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Fqueens~I7e4a6820aca84904752901a63203e1e88e9d3952","project":"openstack/openstack-ansible-os_cinder","branch":"stable/queens","topic":"goodbye_42.3","hashtags":[],"change_id":"I7e4a6820aca84904752901a63203e1e88e9d3952","subject":"Remove openSUSE Leap 42.3  jobs","status":"MERGED","created":"2019-10-14 17:22:12.000000000","updated":"2019-10-14 19:58:48.000000000","submitted":"2019-10-14 19:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688511-1571083128119-14372de2","meta_rev_id":"e8469d0e9734d2eb92c83976876b84dfeb2ba652","_number":688511,"virtual_id_number":688511,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 19:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-10-14 17:24:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-10-14 17:24:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 17:24:19.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-14 19:58:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41062b52f3dc01d344688da0415e7c56a1e3dd10","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-14 17:22:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c1aa97ea881fc086aa266668e96d3d8997420c8","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-14 17:24:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7838643337231092a3a950994067c95ca87af11d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:36:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f68d04a0dcf54ac49fa8e24e110b2472 : SUCCESS in 5m 38s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eabac7ff67aa46ebaadbe23c574e2fcb : SUCCESS in 9m 10s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3ec909b6bbd4412daa805b14e155c13c : SUCCESS in 57m 56s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c0d03ed2983743b89690ff49ab4923b2 : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b0657ad54f3105626986d9500ee3b53caae1addd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:36:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51c3fa98721b059c1c3f412ad6277fd4fede3471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:58:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/602ef42ccf434a9396ce6838526575ae : SUCCESS in 4m 18s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6fa54e09fb894f11a990c808b54c04ad : SUCCESS in 9m 56s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8be74436bdfb4fc48272877acde5850f : SUCCESS in 1h 19m 36s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/271ce92f70db4584a510620bdf24a8d8 : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f7b426ffe96fe110fbdeb95ba95e27c30069e835","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:58:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"881473d0f6f54ff8030b8400db183731d9fcf483","revisions":{"881473d0f6f54ff8030b8400db183731d9fcf483":{"kind":"REWORK","_number":1,"created":"2019-10-14 17:22:12.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/11/688511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/11/688511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/688511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/688511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/688511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/688511/1"}}},"commit":{"parents":[{"commit":"b180454346cc84dc2e70a14d95ba2ea5614928e8","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b180454346cc84dc2e70a14d95ba2ea5614928e8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 17:22:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 17:22:07.000000000","tz":120},"subject":"Remove openSUSE Leap 42.3  jobs","message":"Remove openSUSE Leap 42.3  jobs\n\nopenSUSE Leap 42.3 is EOL since June 2019, let\u0027s remove remaining jobs so\nthat we can remove Leap 42.3 from Opendev infrastructure.\n\nChange-Id: I7e4a6820aca84904752901a63203e1e88e9d3952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/881473d0f6f54ff8030b8400db183731d9fcf483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/881473d0f6f54ff8030b8400db183731d9fcf483"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
