)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~694295","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~Iafcaac9c1f67daf06dfca5a3a49e96b3bc72a11d","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/opendev","hashtags":[],"change_id":"Iafcaac9c1f67daf06dfca5a3a49e96b3bc72a11d","subject":"Replace git.openstack.org with opendev.org","status":"MERGED","created":"2019-11-14 13:36:08.000000000","updated":"2019-11-18 19:28:17.000000000","submitted":"2019-11-18 19:26:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694295-1574105208992-81677534","meta_rev_id":"74e0f3c52a02ead82c245b9a83096aeb8ff78e2d","_number":694295,"virtual_id_number":694295,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 19:26:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-18 16:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-11-18 10:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-18 16:47:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 10:09:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-18 16:47:58.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-11-18 19:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c1fdf95d7cdbd5e042ee668c63090f6686d592d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 13:36:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b8fce1d770f6bc6dd50aa070645d881f8bc64b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 16:16:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e2111bd51f8649dd81c61965286a15b0 : SUCCESS in 7m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7ca95ca4485b4ec09bf76559b9ab57c4 : SUCCESS in 1h 47m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/75cedf97666745739c4ade36874a3c0f : SUCCESS in 1h 43m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/03bab5a04d4e4f7faa3e6703bb431d4c : FAILURE in 27m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a02bcebc95d4bee875b80a82dae4e68 : SUCCESS in 1h 25m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eb2d204007394c07bc1d3f9103885ff9 : SUCCESS in 2h 37m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f753ee6f59954d3eada901d7927e25fd : SUCCESS in 1h 42m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c984aa7366924b729514d131afc3111d : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e817586f2c01417fb6a9ee500a9602a1 : FAILURE in 17m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0760547aa88348f9a544b91a755d3e1c : SUCCESS in 1h 04m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec41c62795604c79b3beec3370d7655f : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af8d9dd355d74d16b390b2ff05bf706f : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"12f581ff241ab89e1ae257711578126695d7fee0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 16:22:17.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9d1211c8cad3d8e9ea52203e08084075752381c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 18:48:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7020cf040b534888bba88fd824dfbc1c : SUCCESS in 7m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/571367204c004a309cb9e30dda75168a : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/07e98e4d9b784570a0f011dd7a6cd60c : SUCCESS in 1h 38m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fa47f1992c6d4afcadb831d23b9836e6 : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e0d9e9582224ddbaafcf2c94b9f52ba : SUCCESS in 1h 12m 47s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/24586d6d8587420abdc36f1690ce4dc3 : SUCCESS in 2h 23m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/461dc5fc1da84ac694b4803e4d6bb1cf : SUCCESS in 1h 55m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b722f27618f24f0583fd5ff532eee5cb : SUCCESS in 1h 05m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ca6e759b4ed5471fab6f008a7816ad6a : SUCCESS in 57m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79ce223f5bfe46c79062e47d77aa4e22 : SUCCESS in 1h 22m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa472f1e6604bca8064c118ebcda749 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f755048e0884688bf3dc6fdec22214b : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b9cd853ffe492a878b261fcd2d8cabbe7c53fb10","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 10:09:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"057ab73ff4519555a2fc4e3ab0b4c47852d1fc20","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-18 16:47:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"51f9c674220215ab7bbea5ee855680fbbb5315af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 16:48:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d593d4c06afa395c0054aa503dd84308de6288eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:26:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"55e3fb7fb6fb98f0ad82c4a469efce356281e293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:26:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8a0dbe23446b48198538154f599dc1e7 : SUCCESS in 5m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2abae23e203642e6bd1e6d6af316b04b : SUCCESS in 1h 27m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3dfd6dbd1a8f4cfaad58e1c66908b515 : SUCCESS in 1h 45m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b3534854b8cd4c65a544c5dc0b07755d : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/387ea124717e48e0829dced2231f15d3 : SUCCESS in 1h 11m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/712ba87ef8964a8dbe21600c7d1b7ae5 : SUCCESS in 2h 32m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/130d02daa8bf466d8801277b3e9d12bd : SUCCESS in 1h 50m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b764f028605649d6b27641d8bf3ddf09 : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06c50c4e7d084072a0403a47b0d10ff5 : SUCCESS in 1h 04m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2747057fa6b46b8902ed0386abd140d : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87d30bb5203b49309cd7c7858e6620ca : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8abfac28cf59400862670ede20d66eb26a0991d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 19:28:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d436fbeb1ed54b6183d6147c6e96b760 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/394700894c9b4dcd9cdfa1e8dfbd5b40 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d45ccefce071ff7c725a235f3828583c1d57c2f","revisions":{"c5d508d8eeeeb77ea1d9e3ebdfa48520653f3312":{"kind":"REWORK","_number":1,"created":"2019-11-14 13:36:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/95/694295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/95/694295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/694295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/694295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/694295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/694295/1"}}},"commit":{"parents":[{"commit":"453166431ad317a7b5cfd814667a074e38c9b265","subject":"Fix typos in the yml files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/453166431ad317a7b5cfd814667a074e38c9b265"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 13:35:55.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 13:35:55.000000000","tz":120},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\nAlso we change upper constraints url due to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Iafcaac9c1f67daf06dfca5a3a49e96b3bc72a11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/c5d508d8eeeeb77ea1d9e3ebdfa48520653f3312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/c5d508d8eeeeb77ea1d9e3ebdfa48520653f3312"}]},"branch":"refs/heads/master"},"1d45ccefce071ff7c725a235f3828583c1d57c2f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-14 16:22:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/95/694295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/95/694295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/694295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/694295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/694295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/694295/2"}}},"commit":{"parents":[{"commit":"453166431ad317a7b5cfd814667a074e38c9b265","subject":"Fix typos in the yml files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/453166431ad317a7b5cfd814667a074e38c9b265"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 13:35:55.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-11-14 16:22:15.000000000","tz":0},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\nAlso we change upper constraints url due to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nDepends-On: https://review.opendev.org/693841/\nChange-Id: Iafcaac9c1f67daf06dfca5a3a49e96b3bc72a11d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/1d45ccefce071ff7c725a235f3828583c1d57c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/1d45ccefce071ff7c725a235f3828583c1d57c2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
