)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~709483","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I7015c56654c122cbbaafef42d99f09d4021a6249","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I7015c56654c122cbbaafef42d99f09d4021a6249","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-02-24 13:08:58.000000000","updated":"2020-02-25 13:48:55.000000000","submitted":"2020-02-25 13:46:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709483-1582638382610-a663f06a","meta_rev_id":"34101674a3e360bf487bc75a65ce046b95b5e395","_number":709483,"virtual_id_number":709483,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 13:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-25 11:56:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-25 11:56:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 11:56:32.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-02-25 13:46:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e852cb0c46d8366b140d26069729c521076f617","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 13:08:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cee2d2f2f6d4e3c7b8a9519b8e167f0caf0534d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:57:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9bf1f189b62e422f97f72d5b7c3feed2 : SUCCESS in 7m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7b960fd820664e0d9f052e6e6b32ca08 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bc1ef01a27a0447a8220e8354b45ea0d : SUCCESS in 1h 37m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/eb40ce3d9422410fab6ec8987c67416a : SUCCESS in 1h 34m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ebbed8289614b0b83c92de49ff2db94 : SUCCESS in 1h 10m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/64ed8c4dfcc54204a649ac425d56d5b2 : FAILURE in 1h 15m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ba4f66867a14e4ab2c63befdf051f8d : FAILURE in 1h 01m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8ee38b29fb6e4d56b28241f705ddcd43 : SUCCESS in 1h 21m 35s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8bd13e03e66a40d7a85f53e229ea67b2 : FAILURE in 31m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25e0a7c54596425585a353a55e730d19 : SUCCESS in 52m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fefa00b299834df4b67de1ad788173a5 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb5e1f9746c44729ad0aaf2a9ccb63d5 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ba8c59d5765cb67a53ad1ce40fc145ee60210db9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 08:52:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"47dcb74f2383c459c0df6d9aaae95f804ecbeae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:54:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0bca901817f1426e8b2484063e216a76 : SUCCESS in 8m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/047d3bd6265f4904b260aeea25532582 : SUCCESS in 1h 08m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fc4ba67020e2481388f654138cf415ef : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ba4e7215b58a4ac29c40fd13c82964cf : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e88d4d4e329d48dfb0784d5c1377ff47 : SUCCESS in 1h 10m 30s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c6f6c87567ba4dc5ac2b9fce594ca137 : FAILURE in 1h 36m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebe29bb4cbf942b89a17e79ce39181d8 : FAILURE in 1h 05m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e214865573534255ae89bd4d67e06393 : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e0f971386dce4fdbbd365375f2ba427f : FAILURE in 33m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e232b968ca149438eb9ecd396babb5a : SUCCESS in 59m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26c681e2eaa4bf88d2b0f071a031bfa : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6368838a809420698a95a86eee7d579 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6c50f18a6aabcce227f695ba63aefe06ffc9a26a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 11:56:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a5e0a08d84111c48e13cacd2f686574255f65da8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:56:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"022157a5d8ac3a699969914bf6a7c5bc25dfc734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:46:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/59fa4258a5774c6ba877f389ce79834d : SUCCESS in 10m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6c09e71d45974870a348d04baa87bd04 : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5f7d3382802a4cafb3106870d8803ee7 : SUCCESS in 1h 31m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/27fb382c663046cfb9c387a775484cdf : SUCCESS in 1h 41m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84819255d6ee4d77b527a19fa792dc42 : SUCCESS in 1h 15m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/213a7afdca2f48bf9869bb23d9c2cd80 : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a3c9b88c0004ce9b110d9079d2dc40b : SUCCESS in 58m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa66d759cc14bafaccfd0a121869826 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f60a49ee9854254a6745884aab730b7 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0173ae57a6d1dfc9874589ebdc57b0e833bb2dd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:46:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c3e96a6d5b198069e45f69d97c2a8dd31f8c6db4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:48:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf3ea889863f452a877c2f6544cc171c : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92eafdf8b0f84b12897568bbedab5105 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"048dc7435e1cb974b7288cb67d78403c6744a5f5","revisions":{"048dc7435e1cb974b7288cb67d78403c6744a5f5":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:08:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/709483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/83/709483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/83/709483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/83/709483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/83/709483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/83/709483/1"}}},"commit":{"parents":[{"commit":"bcdf5843d5b5d7c24800cc00092ede8925c04712","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/bcdf5843d5b5d7c24800cc00092ede8925c04712"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:08:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:08:54.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I7015c56654c122cbbaafef42d99f09d4021a6249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/048dc7435e1cb974b7288cb67d78403c6744a5f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/048dc7435e1cb974b7288cb67d78403c6744a5f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
