)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~739642","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Frocky~I66a76fdb30ae6afd8a8faf4119bef495600aec50","project":"openstack/openstack-ansible-os_cinder","branch":"stable/rocky","topic":"policy-in-code-stable/stein-stable/rocky","hashtags":[],"change_id":"I66a76fdb30ae6afd8a8faf4119bef495600aec50","subject":"Fix policy configuration","status":"ABANDONED","created":"2020-07-07 02:58:16.000000000","updated":"2020-07-07 04:12:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ad10cb1c16aa1d76439d416223ca4a32aced3493","_number":739642,"virtual_id_number":739642,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-07 03:41:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-07-07 02:58:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 02:58:16.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2020-07-07 03:41:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bb212a72e15af5528d9dffffda85f6db412050c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-07 02:58:16.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"00dfd156e8b9db23422455ad1210f3d7d7b575b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 03:41:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b5dc0894ad264a9bb9098d19eb79ab24 : SUCCESS in 6m 33s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/515f8d41557d442eb70fc13bd6d23ed6 : FAILURE in 38m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/293713dfbf3542e4badf41d90d1b97a2 : FAILURE in 8m 08s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f8e71bd4c0b14047aa9841eb0102699f : FAILURE in 36m 25s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/01d36093d5494abe9c84da5b60b27a30 : SUCCESS in 4m 13s\n- openstack-ansible-functional-distro_install-centos-7 https://zuul.opendev.org/t/openstack/build/0b53af901c804661b9974b31b2b32065 : FAILURE in 36m 03s\n- openstack-ansible-functional-distro_install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0f3974a230c4a6188a1db57b6bb6b21 : FAILURE in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2ff82a7a81cb752069800ccdd73470940e0b308f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-07 04:12:26.000000000","message":"Abandoned\n\nthis can be overwritten with cinder_cinder_conf_overrides","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7cec97e76e96391de7e5ab56c1fcf32c55e3223b","revisions":{"7cec97e76e96391de7e5ab56c1fcf32c55e3223b":{"kind":"REWORK","_number":1,"created":"2020-07-07 02:58:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/739642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/42/739642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/42/739642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/42/739642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/42/739642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/42/739642/1"}}},"commit":{"parents":[{"commit":"e01c225a8d85519da158aba3df96843ede4ed3a3","subject":"Remove openSUSE 42.3 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/e01c225a8d85519da158aba3df96843ede4ed3a3"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-06-15 00:36:07.000000000","tz":-300},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-07 02:58:16.000000000","tz":0},"subject":"Fix policy configuration","message":"Fix policy configuration\n\nSince cinder moved to policy-in-code, oslopolicy-policy-generator\nmust be used to compile a policy file that can be distributed to\nHorizon for policy settings. For cinder, it will only output its\ndefault policy unless cinder.conf contains an oslo_policy\nconfiguration section pointing to the policy config file.\n\nChange-Id: I66a76fdb30ae6afd8a8faf4119bef495600aec50\n(cherry picked from commit dbaf649e0c3992f9ab6495ad2f8f55a0a97cd3ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7cec97e76e96391de7e5ab56c1fcf32c55e3223b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7cec97e76e96391de7e5ab56c1fcf32c55e3223b"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
