)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~750491","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I97b50cfe3140d0a0f77f6e562cb144cb3d36bd12","project":"openstack/openstack-ansible-os_cinder","branch":"master","hashtags":[],"change_id":"I97b50cfe3140d0a0f77f6e562cb144cb3d36bd12","subject":"Use cinder_service_setup_host for qos and types creation","status":"MERGED","created":"2020-09-08 20:50:16.000000000","updated":"2020-09-21 15:20:03.000000000","submitted":"2020-09-21 08:55:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750491-1600678551562-0e5654ba","meta_rev_id":"3dd8d2030c32e993ca00cf482371d7e2f15f7a16","_number":750491,"virtual_id_number":750491,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 08:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-09-21 07:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 06:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-09-21 07:23:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 06:51:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-21 07:23:14.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-09-21 08:55:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"112f711a43b85afe9d8258b1e8ab09ec9a18971e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-08 20:50:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bc07acc6c0b12a949e966d13f748a666a8869c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 23:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7f3ff96b7faa4b3d82895c6c91d92842 : SUCCESS in 9m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1f70464e44ef4cfb9f44304fa4c07868 : SUCCESS in 1h 08m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c48485c0bafb44d5a52891953dd76ea2 : SUCCESS in 1h 21m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cad380180c234f2587c94bd3b54e6f37 : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57e0454fdc404e3bac8337b2262bab0d : SUCCESS in 1h 06m 14s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f7fe5fabe36a4bff9ec3c40c8cad7372 : SUCCESS in 2h 00m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68d004b77c494166b0fd074085eec572 : SUCCESS in 1h 30m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e91881cce455452cb349f699af297a4b : SUCCESS in 1h 02m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba08336a98e04a89a67f459ca77fe218 : SUCCESS in 58m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7cb85c77394491803a58492f39a88a : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"cafeef1f56f74fe140c296412ce76e4c6ccbcf62","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-10 07:32:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7374620b7eaad43b59d3ab6c9dcf520485999618","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-10 08:42:58.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f30d5c39ca064d534289de63e329101ef4d601a2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-10 08:58:45.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a6007e3c688e93a44208af2fb40353eadf1c8d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 10:51:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e0d2ed0d72c1497c9206640c5799b59a : SUCCESS in 9m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/011269f884fd49d8ad9205fd7291fdd9 : SUCCESS in 58m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/34aa0fde0dfe41d9bc0ac9dd895a489a : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/281c1a6171ff46e8988417c134981c1d : SUCCESS in 1h 02m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fc21fa35c6d4091a95f784dd59e42d7 : SUCCESS in 1h 06m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1cc5d6bf26284338957ba86d9f828bf8 : SUCCESS in 1h 45m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eeeeb8019f44decb1c5d69c8174f70a : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b3b1e25d5f3e4fbca36ffbe628d4b577 : SUCCESS in 1h 00m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfd8cb8f0c5c43d9a58333ce5560bbec : SUCCESS in 51m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b341c0268c04a4b94d8d95a972953d6 : SUCCESS in 17m 01s","accounts_in_message":[],"_revision_number":4},{"id":"c6d436be2ee24a44ae3928397afe420ca24d1feb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-18 06:51:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a723a6269b8d0074b89edc2be6c48d7eb84318e8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-09-21 07:23:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"32c77fe894c0486b16fec8c9af9f4e0189eb7e15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:23:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"59d1be40e7be936e549790de4cddec594877cfd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:55:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8443d2cc2a744e3296a605065bcc794c : SUCCESS in 8m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b8178a591a48498bb470b4a59cd2b38b : SUCCESS in 1h 04m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f12422e7d5b54b3e92aa2f750f4bdb02 : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0d3a8841d0b496d9534ee905b2380ff : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ac5601f5f3e4713b6636658ac8ab7de : SUCCESS in 1h 08m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9fb8f64a79e84eabb27d75ae52f644a7 : SUCCESS in 1h 02m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dea8ce84ee55438895e89197f2d8467c : SUCCESS in 50m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c84e2d755d740caaa77bc6835c4de55 : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":4},{"id":"4f9a3d35d426d4ea32c9e29e823df48a6ca740ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:55:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c66d5c855f7a2a1d70434e0682f7251fdb252e37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:58:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4b0f37a04747bd8987f58ed8a4ce30 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":4},{"id":"3dd8d2030c32e993ca00cf482371d7e2f15f7a16","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-21 15:20:03.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit d01cdadeb99dd9b0ec6a5f5896ca31af2e363c81","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e89b8640657dfd14641990de8b7f9e1dd842fc27","revisions":{"f8edeff6105478b4d74d67dd4c426d82aa798660":{"kind":"REWORK","_number":1,"created":"2020-09-08 20:50:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/750491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/91/750491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/1"}}},"commit":{"parents":[{"commit":"87bbbe869a4cc7717911f1b6fc13018b8259cd87","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/87bbbe869a4cc7717911f1b6fc13018b8259cd87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-08 20:50:00.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-08 20:50:00.000000000","tz":180},"subject":"Use cinder_service_setup_host for qos and types creation","message":"Use cinder_service_setup_host for qos and types creation\n\nCinder api container may not have required binaries and clients in order\nto add qos types and volume types. In the meanwhile\ncinder_service_setup_host is supposed to have all necessary things for\nthe successful creation of the resources.\n\nChange-Id: I97b50cfe3140d0a0f77f6e562cb144cb3d36bd12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f8edeff6105478b4d74d67dd4c426d82aa798660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f8edeff6105478b4d74d67dd4c426d82aa798660"}]},"branch":"refs/heads/master"},"eb021b334dd51c0a3bdfba25470b43c89e95e788":{"kind":"REWORK","_number":2,"created":"2020-09-10 07:32:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/750491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/91/750491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/2"}}},"commit":{"parents":[{"commit":"87bbbe869a4cc7717911f1b6fc13018b8259cd87","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/87bbbe869a4cc7717911f1b6fc13018b8259cd87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-08 20:50:00.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-10 07:32:23.000000000","tz":180},"subject":"Use cinder_service_setup_host for qos and types creation","message":"Use cinder_service_setup_host for qos and types creation\n\nCinder api container may not have required binaries and clients in order\nto add qos types and volume types. In the meanwhile\ncinder_service_setup_host is supposed to have all necessary things for\nthe successful creation of the resources.\n\nChange-Id: I97b50cfe3140d0a0f77f6e562cb144cb3d36bd12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/eb021b334dd51c0a3bdfba25470b43c89e95e788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/eb021b334dd51c0a3bdfba25470b43c89e95e788"}]},"branch":"refs/heads/master"},"ba9ccb45d185d912464e14cb9292cba3e46d3060":{"kind":"REWORK","_number":3,"created":"2020-09-10 08:42:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/750491/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/91/750491/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/3"}}},"commit":{"parents":[{"commit":"87bbbe869a4cc7717911f1b6fc13018b8259cd87","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/87bbbe869a4cc7717911f1b6fc13018b8259cd87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-08 20:50:00.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-10 08:42:56.000000000","tz":0},"subject":"Use cinder_service_setup_host for qos and types creation","message":"Use cinder_service_setup_host for qos and types creation\n\nCinder api container may not have required binaries and clients in order\nto add qos types and volume types. In the meanwhile\ncinder_service_setup_host is supposed to have all necessary things for\nthe successful creation of the resources.\n\nChange-Id: I97b50cfe3140d0a0f77f6e562cb144cb3d36bd12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/ba9ccb45d185d912464e14cb9292cba3e46d3060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/ba9ccb45d185d912464e14cb9292cba3e46d3060"}]},"branch":"refs/heads/master"},"e89b8640657dfd14641990de8b7f9e1dd842fc27":{"kind":"REWORK","_number":4,"created":"2020-09-10 08:58:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/750491/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/91/750491/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750491/4"}}},"commit":{"parents":[{"commit":"87bbbe869a4cc7717911f1b6fc13018b8259cd87","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/87bbbe869a4cc7717911f1b6fc13018b8259cd87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-08 20:50:00.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-10 08:58:40.000000000","tz":0},"subject":"Use cinder_service_setup_host for qos and types creation","message":"Use cinder_service_setup_host for qos and types creation\n\nCinder api container may not have required binaries and clients in order\nto add qos types and volume types. In the meanwhile\ncinder_service_setup_host is supposed to have all necessary things for\nthe successful creation of the resources.\n\nChange-Id: I97b50cfe3140d0a0f77f6e562cb144cb3d36bd12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/e89b8640657dfd14641990de8b7f9e1dd842fc27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/e89b8640657dfd14641990de8b7f9e1dd842fc27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
