)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~750891","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I0aa2867b703abdd0f86d6e235dc45516acd21d74","project":"openstack/openstack-ansible-os_cinder","branch":"master","hashtags":[],"change_id":"I0aa2867b703abdd0f86d6e235dc45516acd21d74","subject":"Add openstackclient for CentOS distro installs","status":"ABANDONED","created":"2020-09-10 07:30:24.000000000","updated":"2020-10-06 14:29:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1f7a6953287d2fe1f701d840f861cb826452fcc","_number":750891,"virtual_id_number":750891,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-10 09:36:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-09-10 08:59:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 09:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"195af7367def8a2394d31500804c91707d3e678f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-10 07:30:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8009572eebb3bf7e66b82693c3023dee1671261","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-10 08:59:24.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5e9b8d9ec6e2effc9f163eea5f5218c09bdb1e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 09:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6d205dfdb063490ab01f986e74deb78b : SUCCESS in 8m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/07a8a4a66bc14767a9216a8d10d4debe : SUCCESS in 1h 00m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/65148ab50e3c4ec9a42b2f4af38d4551 : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77a012977790440fb44f8fb8524e2fb4 : SUCCESS in 1h 05m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbd66e76f2f54d538c12a12697e166e2 : SUCCESS in 57m 29s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3e3f1f937d894630b821d8bf44c2220f : SUCCESS in 1h 49m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d031d5de06d4dc5b819c0ec179a8cf9 : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e60e7379aae45c29229fa69e4a66f26 : SUCCESS in 53m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e154e21b7f3d4ce89bc5a0b60716171c : SUCCESS in 47m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b801799f5294467287a60926c03a057c : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8c5a3d828b68433f34da7d270d4f750541cd63fd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 21:28:19.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"e1f7a6953287d2fe1f701d840f861cb826452fcc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-06 14:29:01.000000000","message":"Abandoned\n\nin favor of 753016","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"751b592a91456d29ba4b1a278c4d52c007687cfb","revisions":{"751b592a91456d29ba4b1a278c4d52c007687cfb":{"kind":"REWORK","_number":1,"created":"2020-09-10 07:30:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/750891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/91/750891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/91/750891/1"}}},"commit":{"parents":[{"commit":"6c68a55a9b10171cff1ba44c075aa1f126f3bbc1","subject":"Use the utility host for db setup tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6c68a55a9b10171cff1ba44c075aa1f126f3bbc1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-10 07:29:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-10 07:29:52.000000000","tz":180},"subject":"Add openstackclient for CentOS distro installs","message":"Add openstackclient for CentOS distro installs\n\nDuring lxc distro setup scenario creation of qos and volume types fails\nbecause of missing openstack binary. For ubuntu\nand source installs this binary is already present.\n\nChange-Id: I0aa2867b703abdd0f86d6e235dc45516acd21d74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/751b592a91456d29ba4b1a278c4d52c007687cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/751b592a91456d29ba4b1a278c4d52c007687cfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
