)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~751260","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~Ibdb078d39b3189c844aed14aa1ae74cb2ce97600","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"run_condition","hashtags":[],"change_id":"Ibdb078d39b3189c844aed14aa1ae74cb2ce97600","subject":"Define condition for the first play host one time","status":"MERGED","created":"2020-09-11 12:24:28.000000000","updated":"2020-09-21 18:50:41.000000000","submitted":"2020-09-21 08:49:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751260-1600678184983-e4987785","meta_rev_id":"55f664ffd16fe9acadbe6cccea8f8903ff84284e","_number":751260,"virtual_id_number":751260,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 08:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-09-21 07:10:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 06:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-09-21 07:10:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 06:43:38.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-21 07:10:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-09-21 08:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8687fedd64bf2b4b2cd983904c9e61ac8f1ac1e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-11 12:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49fe768c2d43799a512237204ca4b14b0f2c4797","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-11 12:51:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d994026833f7ec3cde117a8542cb50376abd55e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 15:13:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41a70c9032dc419aa8e8bd2a52e31a63 : SUCCESS in 7m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5e3670b314ac4d9caa4960bdb3f417b3 : SUCCESS in 1h 03m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c6c0b44e4bb84d9b8a495fedd83ea863 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2921bc33c5a4eb781067754d5055d78 : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21edbdd6cf0e4664b3101d15c682d703 : SUCCESS in 1h 00m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3b12cd9d7af148c58de301362f3d0654 : SUCCESS in 2h 07m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18e2dbcd285f42de929f12d824a08054 : SUCCESS in 1h 32m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c73be652258243649411c7f645412ba5 : SUCCESS in 1h 00m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9380200253be45e8a8892ba319b542d2 : SUCCESS in 56m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc46ffded8d344208bf87b6503c67423 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":2},{"id":"185053ff2fef0818e4b8fa7b4f9c17fb7d0e2f50","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-11 19:15:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f907826f7cd2111a2535f5f5d7da31ad60025a49","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-15 15:59:28.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"5138868c9f615066ee745abc9f66519b40472a8e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 11:54:55.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a14bb01c87897702e2df43f88d928973cbd96603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:18:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cde49a3a174343aea0c7bab46800b04e : SUCCESS in 7m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d8f3732b21db46b0a552c2a2c7072c2f : SUCCESS in 1h 04m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a087786bc6614459afb67d4423707549 : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9988350582904cf388efe7a7a7c82353 : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c59c96b367446feb4a49ee2891c4fb3 : SUCCESS in 1h 02m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/797f3e0a38194d088e44c6bffc7d0136 : SUCCESS in 2h 10m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26f7bfd57ec24f8f82632fa388edff89 : SUCCESS in 1h 38m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2428ec3d020c4beb96db11808e39362f : SUCCESS in 58m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/242fcf94ad8b437aaa0ccc95a0c96a35 : SUCCESS in 55m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96675df5a054018a3ea5be5d1e943eb : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ef13b555e8abaa9760b4693f9900eef48571fb24","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-18 06:43:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"74b4697edd4d05bb0a7253ccf7f8cc27f5faeb53","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-09-21 07:10:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"077ce598efc1d1c23f5461442aaacc6fc42bc2dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:10:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6d8652234a844b58fc9eccbc527b22f7084a454b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:49:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"20bb4b1af83e99461ea57fbf5b6f3060983cd3c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:49:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0534e4a1858d46dba0ce780ca06624c4 : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7f8a9087975c40078c42dc9b8a0c8638 : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a2c45ecc4b9149368984b75a39a1b8a3 : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5173a72507304244ba0efc06128d5f98 : SUCCESS in 1h 09m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9b33082d3f740759822525fc35141e4 : SUCCESS in 1h 02m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7973805c84fa4083af4f605b2b55c4b9 : SUCCESS in 59m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc111642f0fe43269f1c1aebaf5c163d : SUCCESS in 50m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e710def62554a2f93bf91d0c60aed45 : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":3},{"id":"ce3212f7051e437215cb62b8e83f02fbd7bdb237","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 08:52:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fba5e75ca374906ae22401a28daeccd : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3},{"id":"1ed04f76516ae09b0f5a4245256c7b4657272b09","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-21 18:47:34.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e8aced92ebb1b2e482a36649016453e978852c14","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-21 18:50:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7f24f8d34afa5c28a4794add18458bd451747ba3","revisions":{"d39f118771812f892f62bba3ca55b34385977be5":{"kind":"REWORK","_number":1,"created":"2020-09-11 12:24:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/751260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/60/751260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/1"}}},"commit":{"parents":[{"commit":"87bbbe869a4cc7717911f1b6fc13018b8259cd87","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/87bbbe869a4cc7717911f1b6fc13018b8259cd87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-11 12:24:14.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-11 12:24:14.000000000","tz":180},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Ibdb078d39b3189c844aed14aa1ae74cb2ce97600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/d39f118771812f892f62bba3ca55b34385977be5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/d39f118771812f892f62bba3ca55b34385977be5"}]},"branch":"refs/heads/master"},"0f7e548325448a92514ee3681dc2864daf24f68f":{"kind":"REWORK","_number":2,"created":"2020-09-11 12:51:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/751260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/60/751260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/2"}}},"commit":{"parents":[{"commit":"87bbbe869a4cc7717911f1b6fc13018b8259cd87","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/87bbbe869a4cc7717911f1b6fc13018b8259cd87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-11 12:24:14.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-11 12:51:36.000000000","tz":180},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Ibdb078d39b3189c844aed14aa1ae74cb2ce97600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/0f7e548325448a92514ee3681dc2864daf24f68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/0f7e548325448a92514ee3681dc2864daf24f68f"}]},"branch":"refs/heads/master"},"7f24f8d34afa5c28a4794add18458bd451747ba3":{"kind":"REWORK","_number":3,"created":"2020-09-16 11:54:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/751260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/60/751260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/60/751260/3"}}},"commit":{"parents":[{"commit":"87bbbe869a4cc7717911f1b6fc13018b8259cd87","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/87bbbe869a4cc7717911f1b6fc13018b8259cd87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-09-11 12:24:14.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-16 11:54:52.000000000","tz":0},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: Ibdb078d39b3189c844aed14aa1ae74cb2ce97600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7f24f8d34afa5c28a4794add18458bd451747ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7f24f8d34afa5c28a4794add18458bd451747ba3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
