)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~760582","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~Ie5073c1b178109b4daa95be7bec50f9648260eb2","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"ceph_client_distro","hashtags":[],"change_id":"Ie5073c1b178109b4daa95be7bec50f9648260eb2","subject":"Do not symlink ceph libraries for distro path","status":"MERGED","created":"2020-10-30 15:35:47.000000000","updated":"2020-11-02 10:37:58.000000000","submitted":"2020-11-02 10:24:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760582-1604312662240-8d692ead","meta_rev_id":"ef39b8843ad0fd9d9394f48129290d4413dac764","_number":760582,"virtual_id_number":760582,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-02 10:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-11-02 08:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-10-30 17:34:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-01 17:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-11-02 08:42:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 17:34:36.000000000","updated_by":{"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},"reviewer":{"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},"state":"REVIEWER"},{"updated":"2020-11-01 17:14:17.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-02 08:42:59.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-11-02 10:24:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6eb0abdeb7d6a27f595a3173d50cd2b2cd2c88f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-30 15:35:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"296714d7f8cb568d33c20c12194b74f80729f174","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-30 15:36:09.000000000","message":"Topic set to ceph_client_distro","accounts_in_message":[],"_revision_number":1},{"id":"2d54527fcf327aa21f6ad93953c41feac6fe27f8","author":{"_account_id":29239,"name":"Mikael Loaec","email":"mikael.loaec@inra.fr","username":"miloa"},"date":"2020-10-30 17:34:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"54e9693ec8121071ee19e5c92ed3887676c7527d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 17:52:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/67f8b1b6137b44729067a1ae86abefba : SUCCESS in 10m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/42c077df5edd4753b451d0cd950d08a1 : SUCCESS in 1h 13m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8ebfb1149bfb426d9e3515aa7cbb20f8 : SUCCESS in 1h 29m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/baa454a852474cc6b706adf2988e18ff : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/990e57f738744046a17cf26ed6af45d2 : SUCCESS in 1h 07m 34s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/90c5c9029da14efc88dfa15a75467022 : SUCCESS in 2h 05m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7116c970aae54a7c8403b63b33025048 : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3ce2330b88dc4006b1927fd6b7d24a6b : SUCCESS in 59m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a143368f3ed1417d96860419a7af26e3 : SUCCESS in 53m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c1b32d1b7e4e76a217126595d3ca41 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"97b14287c09aea7e972013336542c4c19fdf5c63","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-01 17:14:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8cd994e4b98ddc505baf1c14cfbe01459c9886ed","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-11-02 08:42:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43450c585b7412cd6a0561792dc031ef135437f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 08:43:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"189568f861a83f32109b771dbb3b8de7db22afa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:24:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d2ae8462035043f5b136f175f5994d0e : SUCCESS in 8m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d8e016901a154ff5877fb689f025052b : SUCCESS in 1h 20m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/11bbc504a75c4cf8b1d97f4db937a9ad : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b901d3155ebf4de98f639c9924e32aa2 : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2afe1894b59547deae062bd7a5c9cf4f : SUCCESS in 1h 07m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/92c9c139bb0f44cc89fc36514a0a514c : SUCCESS in 49m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f715023c41e4487b0572e1b14edd261 : SUCCESS in 49m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d78da56e18548a49b17584b953c78d4 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8c69a5225b9caf11118af6559a3075c408e1807a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:24:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"726db223ab3944cb3687c4f8d208d507c14e97e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 10:25:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5139d3d1c64e21a212962d53642a9b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"d9708406e2236447a54da0941961d61b000d43dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-02 10:37:43.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 1cd51fc2403cfffefa946ecfeb4361463ec24ae1","accounts_in_message":[],"_revision_number":1},{"id":"ef39b8843ad0fd9d9394f48129290d4413dac764","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-02 10:37:58.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 0fb5c77f91c9332ad361c3277042b7db77ffbecb","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f005e7c49b56762945619478c7960087aa1fae6","revisions":{"2f005e7c49b56762945619478c7960087aa1fae6":{"kind":"REWORK","_number":1,"created":"2020-10-30 15:35:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/760582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/82/760582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/82/760582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/82/760582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/82/760582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/82/760582/1"}}},"commit":{"parents":[{"commit":"f7ffb6fb9b445052504954e9878ec0f89b50f3a8","subject":"Merge \"Add rsync to required packages for redhat based OS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f7ffb6fb9b445052504954e9878ec0f89b50f3a8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-30 15:35:36.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-30 15:35:36.000000000","tz":120},"subject":"Do not symlink ceph libraries for distro path","message":"Do not symlink ceph libraries for distro path\n\nceph_client role checks length of the `openstack_service_venv_bin` variable\nto determine if libraries symlinking into venv should occur. So for\ndistro path this should be empty string as no symlinking should be done.\n\nChange-Id: Ie5073c1b178109b4daa95be7bec50f9648260eb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/2f005e7c49b56762945619478c7960087aa1fae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/2f005e7c49b56762945619478c7960087aa1fae6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
