)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~762757","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~Id49d07772c3450ab813241030b2a3029dfc5c9f2","project":"openstack/openstack-ansible-os_cinder","branch":"master","hashtags":[],"change_id":"Id49d07772c3450ab813241030b2a3029dfc5c9f2","subject":"Update keystone endpoint creation condition","status":"ABANDONED","created":"2020-11-15 14:32:24.000000000","updated":"2020-11-16 08:21:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f42eca2589259c804f79f9914608c7ad454daff2","_number":762757,"virtual_id_number":762757,"owner":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-16 07:44:40.000000000","_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-15 18:58:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},{"value":-2,"date":"2020-11-16 07:25:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-15 18:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 07:25:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"d04ba356b3e3511ba0347ca23b4bdcc35ed755cf","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-15 14:32:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11a0e8cdbd835f599f2a69226650d4df29231a36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 16:46:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41dde4fd259540cd986fd1ece06a2329 : FAILURE in 8m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2b625d20c4004e28b9c56d6e84195b07 : SUCCESS in 1h 32m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c4be2b0d297848f2bdae91afb0950787 : SUCCESS in 1h 13m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e87bee9ec4148eba7aeb3ed8b29b2fc : SUCCESS in 1h 05m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec84f28617a246c0a917b4f847a44571 : SUCCESS in 1h 18m 58s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/49ad9ec24a3d4fa4b5ef191e28b6d363 : SUCCESS in 2h 09m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ec4db743fdd454dae6b8ed3ea021d28 : SUCCESS in 1h 19m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0d506fcb73e841a682d6ec53c8359320 : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7730ca526aa74bebb6e9fbd0c763520a : SUCCESS in 46m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a872dbfde3e94b29b13924dd84e0001c : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f5eebaffb9d515a32e1a7417be0b2db7cd843948","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-15 17:07:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7e061cb55c8324d05c906729ed84f4df8f632b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 18:58:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a0e8db3de4ff4a86a4854afee0f5f299 : SUCCESS in 10m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f8c5700aba1f44c4a14361ff615eef96 : SUCCESS in 59m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5b895833813444e397d7b8e18edae4ff : SUCCESS in 1h 15m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c715c7d5e4b48b2bbaea9daa04d33be : SUCCESS in 1h 16m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ed78c17d2694ca19a0d46c7f08626ea : SUCCESS in 1h 17m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/985864aebe764e51bd6db2c531c79a64 : SUCCESS in 1h 47m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af515d567c51473fb686ebd1ee05570e : SUCCESS in 1h 20m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/76d844000db4410c9693f9d033b8063f : SUCCESS in 1h 01m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db71d9030cf842589e7cbe846c34605a : SUCCESS in 47m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a518cef0264b85bb4fe148c7bed0cd : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f5711d4fa81e409df9bf71e88dfd2ede35ed6c93","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-16 07:25:18.000000000","message":"Patch Set 1: Code-Review-2\n\nAs written in comment at the beginning of the file, it should be managed in https://opendev.org/openstack/openstack-ansible-tests/src/sync/tasks/service_setup.yml\n\nAlso it seem wider then that and probably worth talking to #openstack-ansible-sig regarding it, as probably module should not fail here","accounts_in_message":[],"_revision_number":1},{"id":"931e7d564b7e551d2d6ab5c5b261524ba8d9e57d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-16 07:27:15.000000000","message":"Patch Set 1:\n\nbecause what we wanted to do here is to remove endpoint when it\u0027s not needed, and this condition wil prevent doing it... Other way is to add failed_when: false but not ideal as well. So module adjustment seems more reasonable here","accounts_in_message":[],"_revision_number":1},{"id":"16d5c804c515838fa9867a9d292da87fd61fd7e6","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-16 07:34:39.000000000","message":"Patch Set 1:\n\n\u003e As written in comment at the beginning of the file, it should be\n \u003e managed in https://opendev.org/openstack/openstack-ansible-tests/src/sync/tasks/service_setup.yml\n \u003e \n \u003e Also it seem wider then that and probably worth talking to\n \u003e #openstack-ansible-sig regarding it, as probably module should not\n \u003e fail here\n\nI didn\u0027t know about openstack-ansible-test. sorry for the inconvenience.","accounts_in_message":[],"_revision_number":1},{"id":"cfc73202e3543ec8b9b4f270415bbc31c190a023","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-16 07:44:40.000000000","message":"Patch Set 1:\n\n\u003e because what we wanted to do here is to remove endpoint when it\u0027s\n \u003e not needed, and this condition wil prevent doing it... Other way is\n \u003e to add failed_when: false but not ideal as well. So module\n \u003e adjustment seems more reasonable here\n\nI get your concern here, I had the same idea as you when I read the bug description, but when I checked the service_setup playbook and the way that _service_catalog and _service_endpoints variables are technically realized, I noticed the \"Add services to the keystone service catalog\" task will take care of service deletion, and therefor the endpoints associated with that service.","accounts_in_message":[],"_revision_number":1},{"id":"a0fa9f37f86c45856132b378eb85c4b910b0f214","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-16 07:51:23.000000000","message":"Patch Set 1:\n\nDue to code here it should not fail https://opendev.org/openstack/ansible-collections-openstack/src/tag/1.0.0/plugins/modules/endpoint.py#L198-L199","accounts_in_message":[],"_revision_number":1},{"id":"f42eca2589259c804f79f9914608c7ad454daff2","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2020-11-16 08:21:59.000000000","message":"Abandoned\n\nwrong place to edit.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9a9801f306572f4fabe2a4e429b0381bbcae7f8","revisions":{"d9a9801f306572f4fabe2a4e429b0381bbcae7f8":{"kind":"REWORK","_number":1,"created":"2020-11-15 14:32:24.000000000","uploader":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"ref":"refs/changes/57/762757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/57/762757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/762757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/762757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/762757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/762757/1"}}},"commit":{"parents":[{"commit":"00a38c6584c09168faad135f10d265ad9c86efba","subject":"Define credentials for nova interaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/00a38c6584c09168faad135f10d265ad9c86efba"}]}],"author":{"name":"siavashsardari","email":"siavash.sardari@gmail.com","date":"2020-11-15 14:27:28.000000000","tz":210},"committer":{"name":"siavashsardari","email":"siavash.sardari@gmail.com","date":"2020-11-15 14:27:28.000000000","tz":210},"subject":"Update keystone endpoint creation condition","message":"Update keystone endpoint creation condition\n\nChange-Id: Id49d07772c3450ab813241030b2a3029dfc5c9f2\nCloses-Bug: 1904029\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/d9a9801f306572f4fabe2a4e429b0381bbcae7f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/d9a9801f306572f4fabe2a4e429b0381bbcae7f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
