)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~768638","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I219a9f29f92d511eb773b115ff12976082ce2cfc","project":"openstack/openstack-ansible-os_cinder","branch":"master","hashtags":[],"change_id":"I219a9f29f92d511eb773b115ff12976082ce2cfc","subject":"Add Victoria release info","status":"ABANDONED","created":"2020-12-29 02:37:18.000000000","updated":"2021-01-26 17:25:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e7f43a53b4b9b9334505a7172bac30fc88800e53","_number":768638,"virtual_id_number":768638,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-30 01:20:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 02:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43b591250b28507048385a23e47732347e8cb183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2020-12-29 02:37:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9be234ee633d4c7e98c59a868cdddb04b8adec16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 02:56:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9b22b16e613345a9941e9a0fdd88a276 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e84e53d614a24e7fad2b49ca2bf41152 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b7e9a1536fd4a6bb4c4b02ad7523dbe : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f8896cc7f13021bae8cf687122e465325aa5fd8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2020-12-30 01:10:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44dfc7aa36025d0893d9c076b3471e7a06dd76f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-30 01:20:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e90ebf583252491ea3ad685712672e01 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674cc948972a4771b057f2f293924631 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3fa6694026b4bf8b04f3a7d10b057f3 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e7f43a53b4b9b9334505a7172bac30fc88800e53","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 17:25:24.000000000","message":"Abandoned\n\nWe are no longer publishing releasenotes for this repo","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94b03c691c8a704c78e78c32f18794db3f3cbdd5","revisions":{"7ebe981085de7fdb07942e85a8d748f928885cd4":{"kind":"REWORK","_number":1,"created":"2020-12-29 02:37:18.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/38/768638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/38/768638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/768638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/768638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/768638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/768638/1"}}},"commit":{"parents":[{"commit":"00a38c6584c09168faad135f10d265ad9c86efba","subject":"Define credentials for nova interaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/00a38c6584c09168faad135f10d265ad9c86efba"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 02:36:14.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 02:37:12.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I219a9f29f92d511eb773b115ff12976082ce2cfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7ebe981085de7fdb07942e85a8d748f928885cd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7ebe981085de7fdb07942e85a8d748f928885cd4"}]},"branch":"refs/heads/master"},"94b03c691c8a704c78e78c32f18794db3f3cbdd5":{"kind":"REWORK","_number":2,"created":"2020-12-30 01:10:50.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/38/768638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/38/768638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/768638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/768638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/768638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/768638/2"}}},"commit":{"parents":[{"commit":"00a38c6584c09168faad135f10d265ad9c86efba","subject":"Define credentials for nova interaction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/00a38c6584c09168faad135f10d265ad9c86efba"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 02:36:14.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-30 01:10:27.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I219a9f29f92d511eb773b115ff12976082ce2cfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/94b03c691c8a704c78e78c32f18794db3f3cbdd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/94b03c691c8a704c78e78c32f18794db3f3cbdd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
