)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~777178","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I299e71fac6efdb89655bd36de72b95400d7000aa","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa-facts","hashtags":[],"change_id":"I299e71fac6efdb89655bd36de72b95400d7000aa","subject":"Use ansible_facts[] instead of fact variables","status":"MERGED","created":"2021-02-23 18:04:40.000000000","updated":"2021-03-04 10:33:25.000000000","submitted":"2021-03-04 10:31:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777178-osa-facts","meta_rev_id":"f4c55f09079b75b04f1c8af353d52bd54c7ffe1c","_number":777178,"virtual_id_number":777178,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 10:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-03-04 09:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2021-03-02 10:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2021-03-04 09:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-23 20:29:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 10:15:39.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-03-04 09:14:42.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"}],"messages":[{"id":"0a77cb024c9924089ed99bcb9585d47dcc82b2f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-23 18:04:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81507878366b2939c024537a6203baca9121c1db","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-23 19:22:28.000000000","message":"Topic set to osa-facts","accounts_in_message":[],"_revision_number":1},{"id":"51bcf099217b1dcf3dadc4419b1f6e7867f06b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 20:29:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9df203f66cbd40adb348555c0a0e417d : SUCCESS in 9m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e8be24a464cf4b02a9fefdcb0acefde2 : SUCCESS in 1h 03m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4483e7a4e2534bcfac1c4efa2ea4acca : SUCCESS in 1h 18m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fcc127215e444758db796e8a8f65edc : SUCCESS in 1h 25m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45e84c53e70c45498ad20e2c1ecccfc4 : SUCCESS in 1h 05m 14s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/21b6e21a4ed24e57882bda48992aa831 : SUCCESS in 2h 13m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca136ae4927348afa814341cb6162384 : SUCCESS in 1h 26m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/21f8114a4e6e40c9ae1e40c521e06a70 : FAILURE in 38m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab09cae9528849aab0806ec1f110860a : SUCCESS in 53m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c5d7c1e6754c57838e183c4dcfe7f0 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"09cebdaf4b97ac86fca9e04913e91249054162ed","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-03-02 10:15:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0ff3b5348e7804e1ee27f4566d8523f79aeceb4c","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-04 09:14:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ccfad21033dc7acc53e27343df151e0ce516ddbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 09:15:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb175f45b9264ec9decbab65d9c0096e9f0f53dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:31:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/65d5b2bfd36143838b1f0f9ad0bea55b : SUCCESS in 9m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/25efb41fc31c46fd803963bb0fd8d9bd : SUCCESS in 1h 04m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e25a751575864cd6ac5ecf3a26a25065 : SUCCESS in 1h 08m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57f00eb40f984985a8455dfd035a945f : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2252a5e62689433793ed18d103184ea1 : SUCCESS in 1h 00m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ea2c2178e82d4b3ba00790196d2a7dab : FAILURE in 30m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f484982cd1154a30b4d441032a3555e5 : SUCCESS in 51m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b86aaf4461474cb5c117ae2f324fab : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9ee5c9c22b3af7def817565eb2d3a18fb7f8a4cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:31:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f4c55f09079b75b04f1c8af353d52bd54c7ffe1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:33:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a3fd2c048b441f82e2006096903116 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4653722ed352e2ac23f9bd175907a552860f856","revisions":{"a4653722ed352e2ac23f9bd175907a552860f856":{"kind":"REWORK","_number":1,"created":"2021-02-23 18:04:40.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/78/777178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/78/777178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/78/777178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/78/777178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/78/777178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/78/777178/1"}}},"commit":{"parents":[{"commit":"6fd2a1a255e4531a3018ec6b3db1424521f04c9e","subject":"Fix cert verification logic for cinder api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6fd2a1a255e4531a3018ec6b3db1424521f04c9e"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 18:04:29.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 18:04:29.000000000","tz":0},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I299e71fac6efdb89655bd36de72b95400d7000aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a4653722ed352e2ac23f9bd175907a552860f856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a4653722ed352e2ac23f9bd175907a552860f856"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
