)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~833238","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I8155264b181d6f21728804ef8260979931597427","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"bug/1961603","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2022-03-15 17:28:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8155264b181d6f21728804ef8260979931597427","subject":"Add configuration option for heartbeat_in_pthread","status":"MERGED","created":"2022-03-11 09:34:04.000000000","updated":"2022-03-16 08:15:20.000000000","submitted":"2022-03-15 17:28:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833238-bug/1961603","meta_rev_id":"78843d14babff4531ef8958a1738415cdc50293f","_number":833238,"virtual_id_number":833238,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 17:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-03-15 13:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-15 15:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-15 15:44:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-03-16 08:15:20.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-11 12:23:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 13:24:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-03-15 15:44:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f77bad8dcae30cc59329041a91e3c069a405aef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-03-11 09:34:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"308b382c9e4834cf3ade6dc4fb7ebe9fafc3064a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 12:23:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a411c03d6ae74d0f8ad9d553ed5d8828 : SUCCESS in 24m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dc1d367e9053461e8aeb834d257986e9 : SUCCESS in 1h 26m 15s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4f74c2a37bf404ba7c557d19a9c148d : SUCCESS in 1h 26m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3414cc2ce9049b5aa68192da4379fb9 : SUCCESS in 1h 31m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9b1937857ff2414c8a55cd6ae1c8101e : SUCCESS in 1h 34m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/098306949ce14811bd9cd7af9546c83f : SUCCESS in 2h 41m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21cd46a7faa949778f9a35bfba8b5294 : SUCCESS in 2h 19m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fb02e9d8af640f49671fac8f2b8a973 : FAILURE in 33m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/714a64f9c1464240a33977cc0edc3862 : SUCCESS in 1h 19m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13b7305b40c4821ad947222da5f2668 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a3db2030d112637e9a50fa6eef46f1c19eabd21c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-03-15 10:42:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf65018c40cd5deb25174195cdcca43029e44e36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-03-15 11:00:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"133ee6fb97e9148bd7442c085232ac8d1f7b141e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-15 13:24:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"083d0b9b1cd12ebad52d0d55aea77fe75f2abac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 13:32:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62a9e3c45c0f477b9c1e4fd1601916e6 : SUCCESS in 17m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b7db23664de44ca0aa4c0e5b08394ae6 : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f97e57c80c84933b09dbfda09c021d0 : SUCCESS in 1h 25m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d375c7d8615e49e59999468b92b209cc : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/080240a1a0e3447486df774445df1b39 : SUCCESS in 1h 31m 01s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f251e61e9354fdc83fbd573e219d8e8 : SUCCESS in 2h 11m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a15d55faa990490897b1f9fac10e0951 : SUCCESS in 2h 26m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e343b4750bf64ef0a0e6bf733a7e85ea : FAILURE in 32m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14ed19e3189f47619e159fd702533558 : SUCCESS in 1h 02m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330378f0a14d4eefbf16970e600962a8 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/627ff15fc2f748198a6f50394d05c3e7 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":3},{"id":"61b014d924173231898dcde518879d9e329025f8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-03-15 15:44:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4aba5fddbf7004cf313e7e4a57c5737305b2653f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:52:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"531eb7eae77fb948b5216b7acaee72d37642c8f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:28:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b285461de852496789dbadb495f02ad7 : SUCCESS in 20m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bb09fb2bc93c4bc49904b4d9480d2ece : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/534c459928f04dac95e85d6a742fe5f7 : SUCCESS in 1h 32m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8be6a429cc0e4789a3f45ac77781b455 : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/44d4906c49bf4f5eb12d9cfa274d1fe4 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89346fe130ca40bdbb8636ea37c8650a : SUCCESS in 1h 07m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbc2697f47443288361feea9c2f070b : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24cfc8bf2fb44ee0a8d99c5685406a39 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d5b683b968096d05bf6667508a7c1f30cd773279","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:28:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8f3b04004b86f25d517548c5b965988b3482545f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:29:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1cc879f53b4353bdb0e4afa99f6f70 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3},{"id":"78843d14babff4531ef8958a1738415cdc50293f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-03-16 08:15:20.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6efa45e2bdc3d478eead0b8d7b179c2ed49f9841","revisions":{"8d6d2597bb7f6ac5a54173d9cda6deeb9b3ce6a0":{"kind":"REWORK","_number":1,"created":"2022-03-11 09:34:04.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/38/833238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/38/833238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/1"}}},"commit":{"parents":[{"commit":"a9eec797b7b015dcd6b7646c78ef42ea6bfe7319","subject":"Remove legacy policy.json cleanup handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a9eec797b7b015dcd6b7646c78ef42ea6bfe7319"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-03-11 09:29:23.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-03-11 09:29:23.000000000","tz":0},"subject":"Add configuration option for heartbeat_in_pthread","message":"Add configuration option for heartbeat_in_pthread\n\nThis configuration option has been observed to result in file\ndescriptor leaks in certain circumstances. A variable is added\nhere so that it can be easily overridden.\n\nRelated-Bug: #1961603\nChange-Id: I8155264b181d6f21728804ef8260979931597427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/8d6d2597bb7f6ac5a54173d9cda6deeb9b3ce6a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/8d6d2597bb7f6ac5a54173d9cda6deeb9b3ce6a0"}]},"branch":"refs/heads/master"},"8ad2ea586d73df4a567025bf74f9f2136d24a5d9":{"kind":"REWORK","_number":2,"created":"2022-03-15 10:42:24.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/38/833238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/38/833238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/2"}}},"commit":{"parents":[{"commit":"a9eec797b7b015dcd6b7646c78ef42ea6bfe7319","subject":"Remove legacy policy.json cleanup handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a9eec797b7b015dcd6b7646c78ef42ea6bfe7319"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-03-11 09:29:23.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-03-15 10:38:13.000000000","tz":0},"subject":"Add configuration option for heartbeat_in_pthread","message":"Add configuration option for heartbeat_in_pthread\n\nThis configuration option has been observed to result in file\ndescriptor leaks in certain circumstances. A variable is added\nhere so that it can be easily overridden.\n\nRelated-Bug: #1961603\nChange-Id: I8155264b181d6f21728804ef8260979931597427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/8ad2ea586d73df4a567025bf74f9f2136d24a5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/8ad2ea586d73df4a567025bf74f9f2136d24a5d9"}]},"branch":"refs/heads/master"},"6efa45e2bdc3d478eead0b8d7b179c2ed49f9841":{"kind":"REWORK","_number":3,"created":"2022-03-15 11:00:12.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/38/833238/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/38/833238/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/38/833238/3"}}},"commit":{"parents":[{"commit":"a9eec797b7b015dcd6b7646c78ef42ea6bfe7319","subject":"Remove legacy policy.json cleanup handler","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a9eec797b7b015dcd6b7646c78ef42ea6bfe7319"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-03-11 09:29:23.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2022-03-15 10:52:22.000000000","tz":0},"subject":"Add configuration option for heartbeat_in_pthread","message":"Add configuration option for heartbeat_in_pthread\n\nThis configuration option has been observed to result in file\ndescriptor leaks in certain circumstances. A variable is added\nhere so that it can be easily overridden.\n\nRelated-Bug: #1961603\nChange-Id: I8155264b181d6f21728804ef8260979931597427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6efa45e2bdc3d478eead0b8d7b179c2ed49f9841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6efa45e2bdc3d478eead0b8d7b179c2ed49f9841"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
