)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~837570","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I351279dab0d90e60573df7db4c80b63320dff787","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"cinder_audit_send_action_optional","attention_set":{},"removed_from_attention_set":{"29605":{"account":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"last_update":"2022-04-14 23:44:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I351279dab0d90e60573df7db4c80b63320dff787","subject":"Add the ability to disable send_actions option in cinder-volume-usage-audit service.","status":"MERGED","created":"2022-04-12 13:50:40.000000000","updated":"2022-04-14 23:46:27.000000000","submitted":"2022-04-14 23:44:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837570-cinder_audit_send_action_optional","meta_rev_id":"0c9222496ed632b79e56775c7d35078463c4aa46","_number":837570,"virtual_id_number":837570,"owner":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 23:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-04-13 11:40:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-14 21:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-14 21:37:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 15:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-13 11:39:44.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-04-13 11:40:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-04-14 21:37:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"2f754883ba32432524e40d81532dbdfde69d7d44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2022-04-12 13:50:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa5666ea65beeedc4a2d1ba6a9a61546001ce5cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 15:57:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e33691ad6e145a39aca1dd282c62330 : SUCCESS in 21m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/896c2fcf017d4c339539ae2e7706b24c : FAILURE in 45m 28s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8c70a4b69e7402789fedcea5d7e6c77 : FAILURE in 56m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf2dd814766c4476a9b82007a2519494 : FAILURE in 48m 34s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/bce541b0499642a58a443f3f32f57cf8 : FAILURE in 52m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1a0023addd442b19af56cc0b117afe3 : FAILURE in 1h 38m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36494f20161945abaffae0a7c4f2fdc5 : FAILURE in 1h 56m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b40dd8d9d9804f3cac74f1cd75458ea3 : FAILURE in 32m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10d3daebfefe4a35b99eeea6968b6b77 : FAILURE in 37m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed31d03e3c548dba143efe7d2d73ccb : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd38b5cc085c4b4282adfbc0c08c01be : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0a5163fef115a99842e0bb1bb77d50b65134fa86","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2022-04-13 05:04:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5018378ee8476dff1ab6f2fc8fc9ecb3ed6681a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 06:17:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da0042a75e3e49e492e89adecc6d7067 : RETRY_LIMIT in 10m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/be08274a4618439d8ebf8f8d1638cdc6 : FAILURE in 42m 55s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/26bcb8bda345456c84549551dec3a224 : FAILURE in 53m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2887197c2ed4863967e5aa0e89a809a : RETRY_LIMIT in 10m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/69ab0536b69145db9354c627d7de1063 : FAILURE in 59m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/505f3548bfb84aa89a4b967663c01592 : FAILURE in 1h 09m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b87a05012dbc4772bf5060ffd4a0f678 : FAILURE in 10m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f504d3095a74a8e9434aaab24c3e487 : FAILURE in 25m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa0c0e7d8b8b4b2b8648e4c86b0298e4 : RETRY_LIMIT in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df6945e5c6f41509e700966b221b5f3 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aef30e043dbd4a87b15a66096995e819 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d82f7bce9f596c083f70ef07d61615ca20c77556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2022-04-13 11:36:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ecc8d24bea2ecfa4b9e39f2115dddb64cbcd4d4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-13 11:39:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9fc7cd76b4335d8d32181cc3fffcda2986c1ece","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-13 11:40:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4d32d656ea916a8b6fc6604a434546a7a082e47e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 14:01:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/422af2de659e4bbc96a9e5d412600e04 : RETRY_LIMIT in 10m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d177c2039c9149bb9b82ba0fab1e8ea7 : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b46dcf9b444845568e3c73024ec6065b : SUCCESS in 1h 27m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fed3273aab294dfda10a6efa77d62c74 : RETRY_LIMIT in 13m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2331c24367544460a6b8f41d2a1ac42c : SUCCESS in 1h 26m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/eeaacd4dfd924f9bbc09639c6f51a1c0 : SUCCESS in 2h 16m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aec40d4f52574fbcaaea54dd1698b33f : FAILURE in 13m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1990b807b64e43d4a402939d03523564 : FAILURE in 34m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f536686f3e8b4a17b8b4840136080c96 : RETRY_LIMIT in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ab5ebd495e4e7585cb7b7c6332ad8b : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/436ddebae91b4c2ca79d67525c87bbe3 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"54bb6d522303fa2f53391249f1a923a57a4a0c54","author":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"date":"2022-04-13 18:33:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2b700818781835fb83bf71c211cf886598d34ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 20:48:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5575c94da89347038c43a2f344609cbf : SUCCESS in 15m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bc3bbe0fd0a949f9ae36afe3c4d9a671 : SUCCESS in 1h 25m 12s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6f117e789d141c7b6060f27b3f2e3d9 : SUCCESS in 1h 24m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6adf49eaa96b49a6be9aadf94923173e : SUCCESS in 1h 31m 03s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/abde0f7f04ca4235a8affb63fc7c234b : SUCCESS in 1h 30m 21s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2d64394cdd14095a4c4c63c826c6f15 : SUCCESS in 2h 01m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/597e3fc8a79e454399bf9c6ee9784a0d : SUCCESS in 2h 05m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/edc2d2a23a49445e816485e61f64acfd : FAILURE in 31m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd74c0270b6540d5ac1754ebc809c257 : SUCCESS in 1h 02m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dde34393ae04f99893e4c451052bf6b : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d778083dd77f4867a6b9ccbb1b8674d1 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a167af631d3aa33c5575da958bd2a4a7f01c8bb8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 21:37:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"398248e6ca9d9dc3856a8911e2b7b49bc28a4d68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 21:40:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"edfbdf98fd16d1c726b5c39ddb536672752d903d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 23:44:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18eac220abea45828733aa6ddc2ad31f : SUCCESS in 16m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9d7ff311ed58438c97764a3fdbb37ace : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7241401684964ac094cc6918db5ce9cc : SUCCESS in 1h 34m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c06cc653864d66aa744e9e6d71e9a1 : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/7cd1494c95a54a82bb891b46341a2c65 : SUCCESS in 1h 58m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7905e6dd80047368e0adb2a85df6975 : SUCCESS in 1h 04m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb67aae66c24c2ab7719758ca9479f2 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e4f9af16375496b9cde5ba0ea09cb1c : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b0c9970047eb517b08d40551164d2687bc3a0e37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 23:44:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0c9222496ed632b79e56775c7d35078463c4aa46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 23:46:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/839f2a6846f044d69050493dfb37f1b2 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"66eb424de5029d92d5bf6cdd3992d9e0c8e2a63a","revisions":{"a07db8d93b911b766739d96088a1bd0bcc2b5d3f":{"kind":"REWORK","_number":1,"created":"2022-04-12 13:50:40.000000000","uploader":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"ref":"refs/changes/70/837570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/70/837570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/70/837570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/70/837570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/70/837570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/70/837570/1"}}},"commit":{"parents":[{"commit":"89f48407c066cf39ec7148c189c18de717dfe1b9","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/89f48407c066cf39ec7148c189c18de717dfe1b9"}]}],"author":{"name":"siavash sardari","email":"siavash.sardari@gmail.com","date":"2022-04-12 13:48:38.000000000","tz":270},"committer":{"name":"siavash sardari","email":"siavash.sardari@gmail.com","date":"2022-04-12 13:48:38.000000000","tz":270},"subject":"Add the ability to disable send_actions option in cinder-volume-usage-audit service.","message":"Add the ability to disable send_actions option in cinder-volume-usage-audit service.\n\nChange-Id: I351279dab0d90e60573df7db4c80b63320dff787\nCloses-Bug: #1968734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a07db8d93b911b766739d96088a1bd0bcc2b5d3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a07db8d93b911b766739d96088a1bd0bcc2b5d3f"}]},"branch":"refs/heads/master"},"66eb424de5029d92d5bf6cdd3992d9e0c8e2a63a":{"kind":"REWORK","_number":2,"created":"2022-04-13 11:36:46.000000000","uploader":{"_account_id":29605,"name":"Siavash Sardari","email":"siavash.sardari@gmail.com","username":"siavash.sardari"},"ref":"refs/changes/70/837570/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/70/837570/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/70/837570/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/70/837570/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/70/837570/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/70/837570/2"}}},"commit":{"parents":[{"commit":"89f48407c066cf39ec7148c189c18de717dfe1b9","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/89f48407c066cf39ec7148c189c18de717dfe1b9"}]}],"author":{"name":"siavash sardari","email":"siavash.sardari@gmail.com","date":"2022-04-12 13:48:38.000000000","tz":270},"committer":{"name":"siavash sardari","email":"siavash.sardari@gmail.com","date":"2022-04-13 11:36:34.000000000","tz":270},"subject":"Add the ability to disable send_actions option in cinder-volume-usage-audit service.","message":"Add the ability to disable send_actions option in cinder-volume-usage-audit service.\n\nChange-Id: I351279dab0d90e60573df7db4c80b63320dff787\nCloses-Bug: #1968734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/66eb424de5029d92d5bf6cdd3992d9e0c8e2a63a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/66eb424de5029d92d5bf6cdd3992d9e0c8e2a63a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
