)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~859337","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~Ifceb2b816199339ec7725bd95cc890595eed95d9","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/test_volumes","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2022-10-05 14:39:21.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-05 17:08:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifceb2b816199339ec7725bd95cc890595eed95d9","subject":"Add nfs and ceph jobs to templates","status":"MERGED","created":"2022-09-26 18:43:52.000000000","updated":"2022-10-05 17:11:06.000000000","submitted":"2022-10-05 17:08:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859337-osa/test_volumes","meta_rev_id":"c331522a165b1958d83eebdb6be0bdebe0f36faa","_number":859337,"virtual_id_number":859337,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-05 17:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-10-05 14:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-05 14:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-05 14:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-04 14:38:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-26 19:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-04 14:38:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-10-05 14:03:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"920e80ebb5f5019a8b4017c241a6ae837d15a7d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-26 18:43:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfa9d5a24c577a8cf68386f6c6f32d2b5428938e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-26 19:20:17.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f267c25fdaf9bd04450b5a94c64117bb95236b78","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-26 19:20:27.000000000","message":"Topic set to osa/test_volumes","accounts_in_message":[],"_revision_number":2},{"id":"c1e3fcc388523b832232ddf7ae74b401d8ebb4f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 19:32:26.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/859339 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"bf2bd58dfe824bdfaf61870d7fc239d3b0fc8da0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-26 19:32:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5bb159cf5a499724830d2a6c0ed28802c38f5d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 21:01:52.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/859339 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"e14eeba3e9bdab68ca53ec7eb749b89c772e7bb7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-26 21:02:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e79a4fb7796f79d4555bee1cc8756c7fe37544c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 23:42:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b88bfd00d3d4dc496abbe05533ad710\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ab55ce5353548d8969f575af7023168 : SUCCESS in 18m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/66af1ee191ba4c74ab10db14fccdf6d1 : SUCCESS in 1h 03m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3a613be4b814988a3b1ede4486e64da : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed58ae2ea34c45aab85995a167cd81a2 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e277badbd7254992ba22b71bd00394ee : SUCCESS in 1h 03m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f15481ff89d4b159e3b4290e1deeecc : SUCCESS in 1h 31m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d228b169e9042199c83529ebc8d2015 : SUCCESS in 2h 32m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0113bf5f7bb54411b5f451f8ad215df8 : FAILURE in 26m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4938a6d4b73b4c6986edfe8a924e9777 : FAILURE in 29m 06s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dc21d739eec4794a2fa17bb753a441d : RETRY_LIMIT in 37m 42s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/985201ad4ccc4ea6a5ec79d57a97af62 : RETRY_LIMIT in 37m 16s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fe82508a6764aeca54e2ea3e8eb8c84 : SUCCESS in 2h 12m 04s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97ca9196ce3449478d70d80b118befe2 : SUCCESS in 1h 27m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5abe6d1b2b80492eacb52529b0b40c90 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"272c309056383650fbd6a2dfb108f72816ce95d1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-27 09:01:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c08d89d7472fe8cee1b8428ce18381b0e0d7873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 09:03:53.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/859339 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"bf0565878d259d662827f0b7a6e2e385d1b4fb64","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-27 09:04:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ad9bf1a018e93170a0c510732631f4cb893bc56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 11:36:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a94e6c75bbdd455288a8b6e4b4bfe064\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1abb932358674e7797db5be71b6fcb4b : SUCCESS in 23m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0db6fb96f4a14ae59fb000d0bec2c01f : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/413fb57184d64e4ea4f2c5c28100a8ae : FAILURE in 1h 35m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdd8f26aa82342738c1acaaca603c5af : SUCCESS in 1h 33m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11b5fd01848341d0a1eb030b686d547b : SUCCESS in 1h 21m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a7a2a261ab84eb5b77084ffcafbdb7b : SUCCESS in 2h 05m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/244ead5210aa4e9dbf1e3b333fff1e6f : SUCCESS in 2h 07m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/597f7c245241433ebcd1bd0c7f8639d9 : FAILURE in 43m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/555d4a8167944871abaeb9106b258b28 : FAILURE in 35m 12s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90a8d97da2b94155ada0d5fdf3e34f7f : SUCCESS in 2h 29m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8b4996c200846408f6974f4f4932d9b : FAILURE in 53m 03s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b589b43d1214862adfa325f1d4b2969 : FAILURE in 2h 08m 14s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4172aa21f4040ca8ed2ae34076424b5 : FAILURE in 1h 51m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0813d4f4f448b399f909a47b38dc71 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":2},{"id":"7fa7bdf364d5badc287d9805644651ed5ea72fb4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-27 14:25:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d9e9103ae5095631a9959a68a1a1ea6f95deb67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 16:58:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/786da75a20b84222b20afd7041f92bf6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/017895ffdc6d46b88b9adf0baa087425 : SUCCESS in 17m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4f780af2e7564ec0ac923a191d3c85e3 : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5d25f84966c47dc8ad27b1befc33980 : FAILURE in 1h 28m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b88b20de24b9410795523233139eb10c : SUCCESS in 1h 28m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc5b640cae8e4e70b0e080eec33d2e77 : SUCCESS in 1h 16m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dfee597f51b4624ad7ff371baf3906f : SUCCESS in 2h 05m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/123f2913f1e14367ba75c580b5f2b8bd : SUCCESS in 2h 07m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/20afb698c8e441bea05a594886daf509 : FAILURE in 35m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cca744e4582740b9b1ebbee88b6701f5 : FAILURE in 30m 18s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/768c3ba657ca4ac994e785a1e4e78550 : SUCCESS in 2h 23m 53s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d68660766b424db4979ba66f62b53114 : FAILURE in 48m 20s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5c0ee4f9b3142c1b1d796902d14ab37 : SUCCESS in 2h 10m 43s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/054e0e1cca69469b87debf293f11b747 : FAILURE in 1h 35m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43611c3770c144e79c96572a21780a6f : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"025a744e1ca1da094e734b393c51ecff817c566b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-04 10:57:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a639dbc51da83865b8d49f4660ea430f145e71d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:08:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d2c33e9e4f4630a41cf00ca9bc4fc0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7f5d24a758a467f9ae6702da5bacab1 : SUCCESS in 13m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e9ba480604ed425d9e8c3ac0490169eb : SUCCESS in 57m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d753309f9aa46eb8e0e538594b87528 : SUCCESS in 1h 04m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d33364f48afe460f994a3e223cc9fcf9 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10e4a2f2b49b4ebf9210221fa8bb23b2 : SUCCESS in 1h 13m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffaa5842f74e4f86902dbee75c19f9c5 : SUCCESS in 1h 05m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/34ca5e42c96c49baa88e5c80f14e1cdc : SUCCESS in 1h 41m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c923568f6c564d2b9c79a59ae51bb70c : SUCCESS in 1h 46m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7431a34a787940febb6a5a1c02fc6737 : FAILURE in 26m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/08d8c45ee1d240119dc68f779edee528 : FAILURE in 33m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be3918766eff498594c4cb8faab4bea1 : FAILURE in 23m 46s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7544f6714526447b969785de7ae2917d : SUCCESS in 2h 03m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e3f110fec734f6299998470c9e245dd : FAILURE in 38m 06s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/593a02ec9a5e49518bb86301763ed70b : SUCCESS in 1h 50m 43s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8128fc302d76400ab2d06a586db9648a : FAILURE in 1h 23m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e759558997594a5baf55a5df3865af93 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"bf4c2cf883065d5e4d9613f0639ab9e31b29ce9b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-04 14:38:53.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"24c7adf685618cec1c448038e00aa08de03a59ed","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-10-05 14:39:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"77bea858ed0ee3f09a644151dd1444c45007e002","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-10-05 14:40:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5337d528beb5a1b9f180636e70af6ae12c64e67a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 14:43:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1d1d180bb973c3c9a38d5eeff7fd0c152b4f8293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 17:08:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/128ee67626ee4cc6b7d28f42e2c2fcad\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e499699d5edf4a1fb3789fdf5ae012f5 : SUCCESS in 13m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/86637e8511b34e75bc5adbd0a57064f5 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2db7789fcdfb463eaea63ca32df15e64 : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f9bbf0d0dbf0405eba709b52be978b94 : SUCCESS in 1h 14m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e669658080cf4652b4b2d92b891f517f : SUCCESS in 1h 32m 10s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba7f59eb0f4c4f28b99c1d0f0d25d361 : SUCCESS in 2h 20m 01s\n- openstack-ansible-deploy-aio_nfs-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d6330d1fe9a4511a85e3d98cca7a14c : SUCCESS in 2h 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8272f9f2caa645c586185cdbb5452a0e : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ee43e0b485c406955c2a7ffe2429cfac6d85290c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 17:08:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c331522a165b1958d83eebdb6be0bdebe0f36faa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 17:11:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b014125751304824ba1c50944b4a7b5a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53934cc20bd42aea2d712463182985a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0d4b99404946d1cbffcd814657d02d9deb9442ac","revisions":{"56e1a5a346f10d21eec7137990faef69059919ef":{"kind":"REWORK","_number":1,"created":"2022-09-26 18:43:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/859337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/37/859337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/1"}}},"commit":{"parents":[{"commit":"13bf575d233cdf804ef6c0adecc606941fa62aae","subject":"Remove redundant vars line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/13bf575d233cdf804ef6c0adecc606941fa62aae"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-26 18:43:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-26 18:43:44.000000000","tz":120},"subject":"Add nfs and ceph jobs to templates","message":"Add nfs and ceph jobs to templates\n\nWith changing cinder code we potentially can break some backends.\nIn order to detect this in time we are adding ceph and nfs scenarios.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/859336\nChange-Id: Ifceb2b816199339ec7725bd95cc890595eed95d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/56e1a5a346f10d21eec7137990faef69059919ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/56e1a5a346f10d21eec7137990faef69059919ef"}]},"branch":"refs/heads/master"},"05ab59648eb880691aca0d40c9fee030574ccac2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-26 19:20:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/859337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/37/859337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/2"}}},"commit":{"parents":[{"commit":"13bf575d233cdf804ef6c0adecc606941fa62aae","subject":"Remove redundant vars line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/13bf575d233cdf804ef6c0adecc606941fa62aae"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-26 18:43:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-09-26 19:20:16.000000000","tz":0},"subject":"Add nfs and ceph jobs to templates","message":"Add nfs and ceph jobs to templates\n\nWith changing cinder code we potentially can break some backends.\nIn order to detect this in time we are adding ceph and nfs scenarios.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/859339\nChange-Id: Ifceb2b816199339ec7725bd95cc890595eed95d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/05ab59648eb880691aca0d40c9fee030574ccac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/05ab59648eb880691aca0d40c9fee030574ccac2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0d4b99404946d1cbffcd814657d02d9deb9442ac":{"kind":"REWORK","_number":3,"created":"2022-10-04 10:57:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/859337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/37/859337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/37/859337/3"}}},"commit":{"parents":[{"commit":"13bf575d233cdf804ef6c0adecc606941fa62aae","subject":"Remove redundant vars line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/13bf575d233cdf804ef6c0adecc606941fa62aae"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-09-26 18:43:44.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-04 10:56:58.000000000","tz":120},"subject":"Add nfs and ceph jobs to templates","message":"Add nfs and ceph jobs to templates\n\nWith changing cinder code we potentially can break some backends.\nIn order to detect this in time we are adding ceph and nfs scenarios.\n\nWe also fix lvm backend for use on RedHat.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/859339\nChange-Id: Ifceb2b816199339ec7725bd95cc890595eed95d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/0d4b99404946d1cbffcd814657d02d9deb9442ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/0d4b99404946d1cbffcd814657d02d9deb9442ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[]}
