)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~884211","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Fyoga~I19c2b03c61f714fedb593da8489e50d3fa08d933","project":"openstack/openstack-ansible-os_cinder","branch":"stable/yoga","topic":"osa/OSSA-2023-003","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-05-30 09:51:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I19c2b03c61f714fedb593da8489e50d3fa08d933","subject":"Define service_user for cinder services","status":"MERGED","created":"2023-05-24 14:04:47.000000000","updated":"2023-05-30 13:46:42.000000000","submitted":"2023-05-30 13:45:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884211-osa/OSSA-2023-003","cherry_pick_of_change":883844,"cherry_pick_of_patch_set":1,"meta_rev_id":"f80adfdefb5878d222834f903bf3b32de9f3ca23","_number":884211,"virtual_id_number":884211,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 13:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-25 21:06:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-25 19:46:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-05-25 21:06:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 16:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 19:46:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"11e13053994a4785fc481e101786a314b35677d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-24 14:04:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6b0d6e62a92a6180aec846866dbe81da9cdb3a1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 16:35:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48be07fd978e4d87a7e27ca740c9d5e4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69df536df80941b689eb5429bec13d3a : SUCCESS in 11m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c93806bd96674ede99ac52b8ae92c2bc : SUCCESS in 1h 07m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ff93a22b9c9407f9f8bf34c4497d9ad : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/974a868c8f4f44239a34357b6d560d25 : SUCCESS in 1h 13m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8270660882c44f448ee1a7903f37d675 : SUCCESS in 1h 27m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a266e221644435daeb031e1dca91aad : SUCCESS in 1h 25m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2d61014f5e04b62a410a0f45820415c : SUCCESS in 1h 13m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/935f5dcc8eab46e7ad416b24bbd521ab : SUCCESS in 1h 20m 05s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e7f30ba21a649448fb86f62b73dbe54 : SUCCESS in 2h 22m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0c0216f40504d7ba18029019140894b : SUCCESS in 1h 52m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb662829fd2a4976939c7837878f7781 : FAILURE in 47m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0137f3727d845d3ab4e826a9ec89fc2 : FAILURE in 45m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93636c556d4948b2af4b1f1b0d754e54 : FAILURE in 44m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c360eb58efad49c2abe478d315e7ff09 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3f1cae2399cbc3930c0d0aeb6c518bbc1e7ec450","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-24 18:33:46.000000000","message":"Topic set to osa/OSSA-2023-003","accounts_in_message":[],"_revision_number":1},{"id":"9935a771fc66b370a364d8bea1266bff31686a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-24 18:50:21.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8c11e20cfacd0bd129cca10e1dd1e2b4e7a142a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 20:52:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d0b515ef1ba4c718b9ce53561218bd1\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca1a84780e9846b986980544154ca44a : SUCCESS in 9m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/405368c2576e4babb596405e1c8e995f : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bf98c7457324be08cf4c71cf5ab0704 : SUCCESS in 1h 23m 34s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/52a3abdc74cc41db9ffd3f7d007f07a9 : SUCCESS in 1h 26m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0758306e533f4b65b8ea3cbaf7481423 : SUCCESS in 1h 22m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9319c6d4e6244c52adc4aa148139973e : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/322d873eb3b443a583b8561d427e7de9 : SUCCESS in 1h 16m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/db47949d037f4f5494f1cf4e3b343976 : SUCCESS in 1h 23m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ec74f09a208464aaa8b0539d3358afd : SUCCESS in 1h 49m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27a4f0148441460e99c81dbd3e07d1f1 : SUCCESS in 1h 43m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/83be322311f34442b421f55fe1752128 : FAILURE in 59m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/737c38864e5b4f6eb90ea0fc1704a6b9 : FAILURE in 39m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3206324377f49438e2302cefc713d75 : FAILURE in 43m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa0ab962a4740099aa9a6758a033677 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"635634e07cfa2fed2dd67326a4813300e142a24d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-25 11:00:40.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"051c0258d8d244b61c23e4015a75917315b0f421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 13:22:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2252016b8d841d2a54946027eec41b6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/117257759c564a1db21aed92f8368212 : SUCCESS in 16m 22s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e39caed18da544868a8d43c92df215df : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/836a3a7715cd48b4b4f1a0bd890cc3c2 : SUCCESS in 1h 24m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed58d6e44dd542578bb695e06b71ce62 : FAILURE in 1h 09m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/08d52dbcc2d947efa74276480379882e : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f90759f1f844592acd20f1fef380a24 : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8457dba7e0734898adfab57a24ed9ff4 : FAILURE in 1h 00m 19s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/69b60b5aa39a4869b272c012e2cf2c46 : SUCCESS in 1h 23m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa302112aa6f4eca873f836e30874c3c : SUCCESS in 2h 14m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da9bd1d9f99a49c696e0b39d22d1f665 : SUCCESS in 2h 00m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e9eebc56d3d4d76ae3eefde582796bf : FAILURE in 56m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c712e8b5e3b145d5b1c879ea1bb92be9 : FAILURE in 33m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5305794764b34d48a687d6cb94686f2d : SUCCESS in 56m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803e1144488b41219aff4771d23ce499 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a9c83d64a9b684b77114f6eba280464a4f78d27d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-05-25 19:46:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dc1015c5233452e58069edd03673492161ff89bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-25 21:06:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"27e1cb0ea7b795f9976ef5e09690cc8e638d2a9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 21:06:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/751db1f807764465ae321234355ef267\n\n\nWarning:\n  Change 884275 in project openstack/openstack-ansible does not share a change queue with 884211 in project openstack/openstack-ansible-os_cinder\n  Change 884213 in project openstack/openstack-ansible-os_nova does not share a change queue with 884211 in project openstack/openstack-ansible-os_cinder","accounts_in_message":[],"_revision_number":3},{"id":"14f56c95cddfcca04219c5ad67ab7ac7934256df","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-25 21:08:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"848326bdfeb60d08140d88131d61bb4f8896dd78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 23:13:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f7f1844f12e4c618b8475a7324e6bf0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fb8cf740daa4f6099bc41cc076e3881 : SUCCESS in 16m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9973dad8708b4860a847367df0f77820 : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/32a0525ef4bd409a8a4f1c742777f144 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/61fb48b746f54f349cad7305b22c307a : FAILURE in 55m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47f74e4656b8467bb6a5d7767895fc08 : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3138e33e0404cdb95ec9ccbf5ad862c : SUCCESS in 1h 25m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e1719ecdcdb4aef878127bad42c938e : SUCCESS in 1h 14m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e7a4456b2f2944f9a26d624f779c1d78 : SUCCESS in 1h 30m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f38939ab82649968f791c9d024a958f : SUCCESS in 2h 02m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20f689568cbf452ba3e74e3ce9894fc1 : SUCCESS in 1h 52m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/22c74ef92de4432b876f4fe8de06fbb0 : FAILURE in 1h 11m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/229c22410c9748419a1f88f93296d0e2 : FAILURE in 39m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad1967e96f3948bdbcd99b0089021720 : SUCCESS in 1h 00m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/532548985bed4eb0a7b08ecbe7959a33 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"6610e2231c10dee842ee375777f641b84234c353","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 23:13:20.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/095c3ee3ad4e4a75ae509066edeb50d4\n\n\nWarning:\n  Change 884275 in project openstack/openstack-ansible does not share a change queue with 884211 in project openstack/openstack-ansible-os_cinder\n  Change 884213 in project openstack/openstack-ansible-os_nova does not share a change queue with 884211 in project openstack/openstack-ansible-os_cinder","accounts_in_message":[],"_revision_number":3},{"id":"4620b7603b3315ad42037914fb7d2d5ff840f698","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-30 09:51:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9b99c56ce2573bc336cbac60f0f3544f053db32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:08:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/314c375ddee34789b96f8d885d431351\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d0607fb6843430e8ea806d5b797df7d : SUCCESS in 14m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/81667d7c645e419aac1138ae99594559 : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f913d492aa15408f877a2e25f538c474 : SUCCESS in 1h 28m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9885ceb919d643ea82a5866e014b33d5 : FAILURE in 51m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/499bae77d300410a9d92de9e4dd4f304 : SUCCESS in 1h 27m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2953884ad4984a6ebb0cc829fbc4e435 : SUCCESS in 1h 25m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86a2bf28009e4e06ad799222db3da5c9 : SUCCESS in 1h 08m 03s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9c73c0d7c43c4a87bb2a794cb7486417 : SUCCESS in 1h 30m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/337e0990a0cc49ab86b880ecac148fd5 : SUCCESS in 2h 11m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/488ee2670a8942bd836362fc68e48ad1 : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/754cc110d9ae4c7287e6e01dc8973eb8 : FAILURE in 59m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6fd6e0901ec641ae8e476bbdd3f9260d : FAILURE in 41m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9d3f078bfde486cb650a18804092574 : SUCCESS in 53m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148678e60ac54e98b0177c6edbf0f632 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"10fe3e1e9802ecb9ae8341120caa917167dacb6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:10:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"091bc1dd3740c48a66e8832a7104ecf5e40be5ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 13:45:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9636921ca7d14ac0bf2d010dbb6e43f5\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/796c987294794378ac7c6d9f3ac09dc3 : SUCCESS in 15m 10s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/738b522e797a401ba1c27b7f3843355c : SUCCESS in 1h 16m 20s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fff8b90a835e410389fe23650ba9d56f : SUCCESS in 1h 29m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c8a2ab59548422289f738de6b927baf : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/488144741e73456aa854c9df9bd86e3b : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/776aa9c0265949f98f0a04c6c079b0aa : SUCCESS in 1h 29m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aec81b8377f249e8b70fa587e209e3c9 : SUCCESS in 56m 35s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bab4086b5af7456b89fb566c03d371c3 : FAILURE in 36m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a1109433a9d4689ad20987c6650e272 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":3},{"id":"fb7e94a3b15885dd52c72e4fe7cb9e9c8674c94d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 13:45:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f80adfdefb5878d222834f903bf3b32de9f3ca23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 13:46:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea95ad77861c4bfc8a6ef106d21c9206\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a53897e3b414d0d8fc8671f42389f4d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4e9cbff1964645c94e4836f3639fd99d4e62f1bb","revisions":{"b008fb5628a8e303e31b77ecacd785ca7c894d15":{"kind":"REWORK","_number":1,"created":"2023-05-24 14:04:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/884211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/11/884211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/1"}}},"commit":{"parents":[{"commit":"e8d373e31830391130dd389f77937c8ab6944ce0","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/e8d373e31830391130dd389f77937c8ab6944ce0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-05-22 13:57:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-24 14:04:47.000000000","tz":0},"subject":"Define service_user for cinder services","message":"Define service_user for cinder services\n\nIn order to cover OSSA-2023-003, a requirement to define service_user\nsection for all cinder services has been added by cinder.\n\nChange-Id: I19c2b03c61f714fedb593da8489e50d3fa08d933\n(cherry picked from commit 1af3003e163e09f917dd124ae874f1bea6fe2c6b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b008fb5628a8e303e31b77ecacd785ca7c894d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b008fb5628a8e303e31b77ecacd785ca7c894d15"}]},"branch":"refs/heads/stable/yoga"},"bf50519289a6d8cad23c74a9a889886e7513f3e0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-24 18:50:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/884211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/11/884211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/2"}}},"commit":{"parents":[{"commit":"e8d373e31830391130dd389f77937c8ab6944ce0","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/e8d373e31830391130dd389f77937c8ab6944ce0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-05-22 13:57:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-24 18:50:21.000000000","tz":0},"subject":"Define service_user for cinder services","message":"Define service_user for cinder services\n\nIn order to cover OSSA-2023-003, a requirement to define service_user\nsection for all cinder services has been added by cinder.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/884275\n\nChange-Id: I19c2b03c61f714fedb593da8489e50d3fa08d933\n(cherry picked from commit 1af3003e163e09f917dd124ae874f1bea6fe2c6b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/bf50519289a6d8cad23c74a9a889886e7513f3e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/bf50519289a6d8cad23c74a9a889886e7513f3e0"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"},"4e9cbff1964645c94e4836f3639fd99d4e62f1bb":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-05-25 11:00:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/884211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/11/884211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/884211/3"}}},"commit":{"parents":[{"commit":"e8d373e31830391130dd389f77937c8ab6944ce0","subject":"Ensure service is restarted on unit file changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/e8d373e31830391130dd389f77937c8ab6944ce0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-05-22 13:57:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-25 11:00:40.000000000","tz":0},"subject":"Define service_user for cinder services","message":"Define service_user for cinder services\n\nIn order to cover OSSA-2023-003, a requirement to define service_user\nsection for all cinder services has been added by cinder.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/884275\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_nova/+/884213\n\nChange-Id: I19c2b03c61f714fedb593da8489e50d3fa08d933\n(cherry picked from commit 1af3003e163e09f917dd124ae874f1bea6fe2c6b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4e9cbff1964645c94e4836f3639fd99d4e62f1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4e9cbff1964645c94e4836f3639fd99d4e62f1bb"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
