)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~887059","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Fyoga~I711241af8f7520b97f2b1cafd1406ff705fb78a6","project":"openstack/openstack-ansible-os_cinder","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-09-28 17:28:42.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I711241af8f7520b97f2b1cafd1406ff705fb78a6","subject":"Use v3 service type in keystone_authtoken config","status":"MERGED","created":"2023-06-27 15:09:49.000000000","updated":"2023-09-28 20:53:28.000000000","submitted":"2023-09-28 20:52:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887059","cherry_pick_of_change":886497,"cherry_pick_of_patch_set":1,"meta_rev_id":"c1a7a7f42a850cd67bb72f210be6675eed5e8108","_number":887059,"virtual_id_number":887059,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 20:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-08-29 19:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2023-08-30 10:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2023-08-30 10:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-27 15:09:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-06-27 17:37:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 19:00:26.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"a0c5a0671b6a23ed9fea1d82a18ee44a8f2f3dc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-06-27 15:09:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"db66e00b682fa0287529df0291c110975841ec05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 17:37:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e2bbdfec4a84e248282a4a865b325db\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8100ee0ba2a462c8560f58b976b1238 : SUCCESS in 12m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d3ef03b68c264d1b9d57d6ba7345a92b : FAILURE in 1h 18m 21s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/11233c461f1b4932bfe50341871d4cb2 : FAILURE in 1h 36m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/403cd3bb33064f13954d413ba38c9a71 : FAILURE in 50m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e768b24df33b4a7faa73e0c416de4c88 : FAILURE in 1h 25m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4180a98d29d2491db0bbf3af7eb826cb : FAILURE in 1h 37m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/741496fd4f4040fc86745b520d01678c : FAILURE in 1h 22m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b5572d52db2644bc9e2fc3a0c8c44a7a : FAILURE in 1h 26m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/21bee9efbf844412b4ad7b1552d6cd67 : FAILURE in 1h 04m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dda3c7ed7ca4767aebfcc6fcd918ac4 : FAILURE in 49m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d35f1789a4da4ce48ef5063973436a17 : FAILURE in 1h 01m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0c5cfe14d114747b4b00ed216b079a6 : FAILURE in 36m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88f3aa997c2949f6bba5292ab63a3a97 : FAILURE in 58m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4f7cefcd66427f996c023ce385013f : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f792e880059506cc0d06d3fed9609e90bd55318f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 10:26:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ec5978cea32a1fbe59bc8b1d9dc62f02fd6739f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:38:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f162e5fb88c4933ab1528675fa12ca5\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1412f7049924eb7a2fdce6ca656f617 : SUCCESS in 13m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3c102077c7e249d484ed8edbb782862e : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7eeeafc04dde482fa17b8e2038c1e43c : SUCCESS in 1h 39m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/545407d277a0443696e47604837ac4a8 : FAILURE in 1h 02m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1a177b716614762baf4312c71f5be49 : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/830547b043b3419ab5441c7734dc95c3 : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a79e15d02ed24624b8eb8607cab1bf2c : SUCCESS in 1h 18m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/0cc3b392d8444169ae7d9aa6d2342a7a : SUCCESS in 1h 37m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6678c26c8e84b699a47565c0f9dd824 : FAILURE in 1h 05m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91e24ddd770a406cb77bbf3e1c06f81a : SUCCESS in 2h 02m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b427374f10b4b108762ce323f49f4c2 : FAILURE in 1h 04m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eedc1535c600424d8900c2c4a85eab2f : FAILURE in 34m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c59adcbdb2954d81b25e9226fcd65003 : SUCCESS in 59m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85e65db0a974d288009bbc2fa98bcde : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4f6d281336e0d2e46e4da482439eb5c72644a1be","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-08-29 19:00:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e76581c53e230d3c3a22401d5d87f1ac752c4a12","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-30 10:05:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b7a82c34301267961a5b52767de23c7b7a928d9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 10:06:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"654cac3b14917e0d3a8c38cdb95a78d766b8b719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 11:37:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c56fc7578d94cd48c70c9e1502b7901\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a85cf5d084c40a3bff9212ab28280c7 : SUCCESS in 12m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ad0395b8d790489b8318dedbee6ddb1e : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/29898c08dbb944a6a1219079a9ea85bd : FAILURE in 1h 26m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3afb3a1c5e044604921566059ef9b282 : SUCCESS in 1h 19m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50e372693d3c432b9394aa0c107b29b5 : SUCCESS in 1h 24m 40s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/fc4793fca07f44848e2b71a36ff0a3b1 : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8df8a58b5e648c49e90824af6121595 : SUCCESS in 59m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5387b07e5264450f966226d16f52fea7 : FAILURE in 41m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78407a76cc345e5a15ad8ad1a96f437 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"19011287ee6c590bc67ce5ae7540a3e6cf331122","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-31 16:26:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5476ba2c780f404040d01eeb75a2ff9df943d73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 18:24:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff050cb74d2b4a1ebdf75f23bc02e72f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cbe69327629496792d3843f17b864b5 : SUCCESS in 16m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8708608fce03424680816d82e46592d3 : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/df9268ba436f4044873f4f34969fa8c8 : FAILURE in 1h 23m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fb155fd6f724986bc2f8d3a94c37471 : FAILURE in 58m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/121bea603f4b4865af237cdae1a4ac9a : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4a3cb5d338c4672ad00625d5392e8e7 : SUCCESS in 1h 21m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84cb26194ab8431db7a6f1e48558e597 : SUCCESS in 1h 14m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f3b5e66e221a421aa835b79f2499c840 : SUCCESS in 1h 28m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a2b79e1033541119888e08305911dd4 : FAILURE in 57m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f6648e730cb4c8db61b7cbf16d2b0b5 : SUCCESS in 1h 55m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0515fa6153014739b766f0babecb4e19 : FAILURE in 1h 01m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7eb3d6abaae84f2bb2e377bc8535f1ff : FAILURE in 44m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e17db46195384b87977e17047ad238c3 : SUCCESS in 51m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b364b4af78094bd4b2bea65751c5983b : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f7efb13d53288c90d1d30c969fc17bf4f5a5c8c8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-12 07:49:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea94ae3220c69cfb0dfc97511f26e3e4c02da5d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 09:55:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20d2f359015d45118e06f324bdad0ef2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88355997e67e48e6ab2d1eda63f3862c : SUCCESS in 11m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/35af8742c50c470aa880beb021ac6383 : SUCCESS in 1h 08m 09s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1defd71ccde54de0a1f451c629b6a9c2 : FAILURE in 1h 35m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8591a7ff006b4e83a6d725989fbeb051 : FAILURE in 1h 03m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8dc3197059ff481fb9784f6d3a3afaba : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32018e55a53f48c8adff4f24cd54b213 : SUCCESS in 1h 28m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f7587f35a3c4c069a7b4d2f23b439ff : SUCCESS in 1h 19m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1391ac8275634fa4b4ef9e666741ae8f : SUCCESS in 1h 26m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea08b0acee1445e2ba624e60b15ec4f6 : FAILURE in 1h 02m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dea7cad517a44bd4bd52adeef3e00862 : SUCCESS in 2h 02m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a555291ee74243e4b0b06f110c5594d6 : FAILURE in 1h 01m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f935a433a734e89b070ad1593409992 : FAILURE in 39m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce199655c0284cd89dbd27fcf34f949e : SUCCESS in 57m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/379928f676334513baa6cb03389db9d5 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e31dd6c81dfe8ff7e41ddc27cdd860fbc24cc151","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-28 17:28:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25fb63584218d6e43e7f013ec4b2279928cee20e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 19:14:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d27ccc629ee4a44bd9959b58d0484a2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a31b2fee42bf4a85974315fd598373d8 : SUCCESS in 11m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/13056b02afa34f2a81ea557289a0db91 : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b64195927d89495a874ee1e8b5df87e5 : SUCCESS in 1h 38m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/99a34b30eb4a4d9587b6d70194b6ff3d : FAILURE in 55m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/707bab8ae27a4fe1abb8dff024418260 : SUCCESS in 1h 21m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da097376dbfa4c60b13aa67c9195fe32 : SUCCESS in 1h 18m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6c4d971bced4a39b039a97abbc90833 : SUCCESS in 1h 17m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d5e6119cb6f5456191ac368971df1d63 : SUCCESS in 1h 32m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/159dce7433c9440a92485d53727039b7 : FAILURE in 1h 08m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd4ac0a447744b9ba2f2997bdb3d64f4 : SUCCESS in 1h 38m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bf1dc3542b64db8948355940e97d5c2 : FAILURE in 55m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c65427d7d0c64b0ba46c9929a10d9031 : FAILURE in 42m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a28c737deb141c2adf2ec1226f7b1d9 : SUCCESS in 56m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f83fdedf785548d983ee6d2da6df6bb9 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ac60c048aa3ac484c214cf964c16d32a44ae7e30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 19:15:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a957487371dfcb5edbcdd7e3de188c24cdb38a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 20:52:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20be6a0ef6124abda8bacad8a39a0acb\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8fce1f6277242d89f7b5eb33943e312 : SUCCESS in 10m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2715435319264ebc8d833fefe1c515e7 : SUCCESS in 1h 20m 51s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/07d97e27804d498a9b5bb5d31f494e0e : SUCCESS in 1h 31m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/713029251e2c47aaa41227fc7ea5e0dc : SUCCESS in 1h 28m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34a67475113245d2bf3c2adf153b3373 : SUCCESS in 1h 29m 40s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b21c79d20ee4444bab3bac7f90e71d83 : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91fc6b310bf341ec81c237e16418e316 : SUCCESS in 55m 39s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79e0e9b26f804d378538612e5e7ec8a1 : FAILURE in 39m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba990cf871aa41e6859d3211224721ed : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"982949a81645af4867a1bd35d4fe910212ab95f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 20:52:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1a7a7f42a850cd67bb72f210be6675eed5e8108","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 20:53:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21c866306cd84e079ca854e8a9ddc9e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4a70dc2a6c4dc8bc1ee985fd38e593 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ffb0d513e93cfed050209822d9a413c5d9b3794","revisions":{"2ffb0d513e93cfed050209822d9a413c5d9b3794":{"kind":"REWORK","_number":1,"created":"2023-06-27 15:09:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/59/887059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/59/887059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/59/887059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/59/887059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/59/887059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/59/887059/1"}}},"commit":{"parents":[{"commit":"4e9cbff1964645c94e4836f3639fd99d4e62f1bb","subject":"Define service_user for cinder services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4e9cbff1964645c94e4836f3639fd99d4e62f1bb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-20 14:09:17.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-06-27 15:09:49.000000000","tz":0},"subject":"Use v3 service type in keystone_authtoken config","message":"Use v3 service type in keystone_authtoken config\n\nThe service type in this config section must match the service\ntype in the service catalog, otherwise limited scope application\ncredentials will not work with the cinder API [1].\n\n[1] https://docs.openstack.org/keystone/2023.1/user/application_credentials.html#access-rules\n\nChange-Id: I711241af8f7520b97f2b1cafd1406ff705fb78a6\n(cherry picked from commit 2820fcc1d6f0b48e9d25a6bdf5f78d629efb1f0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/2ffb0d513e93cfed050209822d9a413c5d9b3794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/2ffb0d513e93cfed050209822d9a413c5d9b3794"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
