)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~900401","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~If0873fd38d40cb2317e644e20f5be5334753d79a","project":"openstack/openstack-ansible-os_cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-24 17:53:24.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"If0873fd38d40cb2317e644e20f5be5334753d79a","subject":"Fix example playbook linters","status":"MERGED","created":"2023-11-08 09:01:06.000000000","updated":"2023-11-24 22:37:27.000000000","submitted":"2023-11-24 22:36:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900401","meta_rev_id":"2c3d7363c984b77fdf96d542e4fc12298785decb","_number":900401,"virtual_id_number":900401,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-24 22:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-11-10 09:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-10 15:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-10 15:03:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 11:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-10 09:31:51.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2023-11-10 15:03:42.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"a9b9bce7c6858049c8f0811333f2634562f601d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-08 09:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d39eb24b5f0c2496894eefae3111063b4edfefb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:40:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7dd10a7987c40c1950d6afa182dcbd4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85d076d2e59c4784b64e888069ee918f : SUCCESS in 10m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bd8c354281964f9d84439616d4337979 : SUCCESS in 1h 02m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e9fee5f936b7428a9f9b87ae2748b08b : SUCCESS in 1h 10m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f84109602fac4b4ea1e6011c9766ad25 : SUCCESS in 1h 13m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26bb0f528d73484494b997eaf4bf35e6 : SUCCESS in 1h 10m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62ae6841fe9b4def9f23a6878d1d6966 : SUCCESS in 1h 05m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/612636fd917e43ff801212e3211ffbf8 : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ba6dca87c274bdaa61a529c6f8b60b1 : SUCCESS in 1h 18m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c331528803574eaab12152ecb6f80c31 : SUCCESS in 2h 29m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8c7a19cc9ba41f49d9b9086c44a61ba : SUCCESS in 2h 18m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6cb1723ff6ed446e8c7c180bb1777dd9 : SUCCESS in 2h 13m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f656202365546db90f4fa849f628664 : SUCCESS in 2h 23m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/84395ee05896462190668518df7f1859 : SUCCESS in 56m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/264624cf46ae4565bead29e526a4eec3 : SUCCESS in 1h 00m 54s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/541b05bae6ed4c6993e05c9d17134819 : SUCCESS in 2h 10m 22s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9961dde3fb38432bba6ae6b73f8f784a : SUCCESS in 1h 59m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e29ec9dbce4659b568b508441823fb : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0bfdd46edc0e64b6107033f9aed420535ae756b5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-11-10 09:31:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ccba7d60e5d0df56497abc861cc89553c2446fa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-11-10 15:03:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfb8376087841239af9e2d2f27a2a49301c394f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 15:05:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58ad078ce0809faa8732890d099729ca2488a5a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:19:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9d0284507e0480eb6e33218650376d4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8299e00a199f4b5488b732365d91062a : SUCCESS in 17m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9abb271d36f44f9abe9f8c2beef04415 : FAILURE in 22m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b30bb33085ee4265b69ecfae9efe59b6 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f9513e6e526d47969128d8d4330fd53e : SUCCESS in 1h 24m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e4592bcf522412bb4c73a2657cdaf14 : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/068dc34cd90246bb85316857b0e03386 : SUCCESS in 1h 27m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e659890a177240bea10579bb79d7192e : SUCCESS in 1h 11m 45s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6ec8cab67124bea9327eb135ea22623 : SUCCESS in 2h 01m 18s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d1427f90df7414cb75f09c845168b21 : SUCCESS in 2h 06m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda99a72b7704fe7a71c0eab26b88279 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ad79da4743e5854869a9b67fe92a670ea02feb6b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-20 13:56:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40ef92b763399922b3a15538905694c8dc152e3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 16:34:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9cd1398ab4047418202198f1442d7a4\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62fb844cb9e540818cc6078ba8cd6717 : SUCCESS in 11m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/74aae8b6b1a049b7b4a268c4c7cc5683 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a581e881a09d4fecb8d59f0fd8629982 : SUCCESS in 1h 15m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/484ccab6e84f44fb81d478ed0fae4c68 : SUCCESS in 1h 13m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d4a59dfa76f4434a9cd05b48545d61b : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/348d77737eff450ba727166f68b39993 : SUCCESS in 1h 07m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/370aacd1b69a4a4c8dea3dcf25be25c5 : SUCCESS in 1h 21m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/895aba3707774d2e9fa2065d6e2a2582 : SUCCESS in 1h 17m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68bbb649a96d44a1bbe39819b81579be : SUCCESS in 2h 29m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed7acb8dbc794aaca61f24e7f5bda808 : SUCCESS in 2h 03m 46s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fcf964c30c054fc2bcce7824e38d02c5 : SUCCESS in 1h 50m 05s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ec925fb541a41e38ab4d236b44b6c73 : SUCCESS in 1h 58m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/57b5f5736a194c31bfeb3f9abe9973ad : SUCCESS in 56m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b83612345d84ae191077d607f794a3e : SUCCESS in 49m 10s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c9b4586f074440e936bbb05892bd407 : FAILURE in 1h 05m 20s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a19ba0f101045aebb162c2970128f75 : SUCCESS in 1h 40m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02dec03ac99c4e0aa668f0fad180d02d : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7ea1460413d5a7e602ae9bbdd99b6d395d56636f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-20 17:50:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e60c9ef01769fa4521579517a034ef48e18ec81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 19:51:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/624eb78aa9004fbbbbbe06bce3ed7c4b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/774188689c9f446b8a72dc1bdd8b729f : SUCCESS in 12m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/786062a4cfec4ebda95d2b87edea279e : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9a9a8d009ea4458bab195a4a3ddcccc5 : SUCCESS in 1h 10m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cc6278b4a1b42dcaff8bccd65dcb844 : SUCCESS in 1h 12m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a3663ff6db142c6a86ab8e1deb43f16 : FAILURE in 18m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/593fcd08e6ce4fc39fd300f9d57a54b8 : SUCCESS in 1h 06m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd6aad4178234ba59f5c77040c2c641b : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ae9f3624eba4723bed20055f47ba0a8 : SUCCESS in 1h 18m 22s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/29066d16416a43228a52909456d36b37 : FAILURE in 18m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49b0210aed3e4144ba8bd6c2dbc75cf0 : SUCCESS in 1h 58m 17s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd95a10496f44cc993ddc513afe46deb : FAILURE in 11m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c878b211d26404983df1564546d6a59 : SUCCESS in 1h 49m 44s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ca0accfa35804454b23b081befad8112 : FAILURE in 19m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afba436726c64afbab4123bba4aa1692 : SUCCESS in 54m 38s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbd8a4876446492c8419003330666cac : SUCCESS in 1h 56m 31s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c91fc7dadb174709bd5af08256eb554c : SUCCESS in 1h 54m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3487a349569e4cbba7cf80651016142c : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"91a9adc74dc852f9f7dd5e2b4d1823020a092a8f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-21 14:19:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e28c74c7203137de78ebb53fa4eaa68fc2277359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 16:59:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b1434ba5387468ca6ab8b8cf4464e04\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c29b0ad4a824941aee6901fbd0cae58 : SUCCESS in 12m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0e9c66c28a8649d5b17685c7a3e710f8 : SUCCESS in 1h 04m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9dff99dd5ed345b48d78bb6b3f75e2b6 : SUCCESS in 1h 23m 23s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5ae2d72483b493997cab80b3f85f0d9 : SUCCESS in 1h 18m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91fdde8f4fc1488ba3cfd6d4058989f0 : SUCCESS in 1h 28m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67100b9374864d2d96752eb45c0db3d0 : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/426fb1ed0a4b4887aa5fa178109d5459 : SUCCESS in 1h 33m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9080b74e08f74d3e9a8e9663138d6e81 : SUCCESS in 1h 02m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f82414de51e4e018b10ced5efa45972 : SUCCESS in 2h 31m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d5ca3623328403081367befbcff9da9 : SUCCESS in 1h 54m 52s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dba642765c4d41948ba3dcc80faeec71 : SUCCESS in 2h 26m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec419c1cc87c48c1bc88469e1c3e4fc1 : SUCCESS in 1h 49m 42s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6aba7123341745d6add7b8861c043b97 : SUCCESS in 1h 00m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4e2485c6d14446a86de98ed99734331 : SUCCESS in 56m 59s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72d28f7e19a542ad9b42fb056c72c335 : FAILURE in 2h 03m 57s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a6be6b1916b4833921e32f08c0f5e93 : SUCCESS in 2h 04m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7142493124e4afa83ca8e6aa82089ca : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0277da1385dc67df718374a7f52bb4a9689ce042","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-24 17:53:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb9e9793d31690bbf637fdb102987ce49d163750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 20:14:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7baa2c30631f4fb185f2f95c07534f55\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8922d156835245f6b3eb48d7e7291876 : SUCCESS in 9m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/77507f2e72ff4f04a68c843c96a77cfe : SUCCESS in 1h 15m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/eaac5f88ddea4904bc5125d6b37d0cd2 : SUCCESS in 1h 11m 49s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/632ebbb6f43c463fafe79c694493788f : FAILURE in 1h 07m 41s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1254876a1b784794980a9e412fef66a4 : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97f0a52cf61841dba4f3b60f9a249716 : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2d97a97196a4bb8a0c387a4c52437d3 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aff48aa5fb741d1b9313340b2b9ef1e : SUCCESS in 1h 14m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3cd1768f74544784a815305830b21a02 : SUCCESS in 2h 18m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f2f5f9815c043eab08098416e353f32 : SUCCESS in 2h 06m 52s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a60ecfca184543179c7d22a59f5ec137 : SUCCESS in 2h 14m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e94a2a7b2464075a1b954d8b3954d31 : SUCCESS in 1h 40m 37s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/775f5bab9b074c998fae56bdcac73057 : SUCCESS in 59m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27714935fa07489eb9bb66d43ca9765c : SUCCESS in 54m 55s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5be3d1d3622842b995e5bc90fec19032 : SUCCESS in 1h 52m 56s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d96b10affde74e2cb65c1857cbcdc49b : SUCCESS in 1h 40m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7008812f8de24811a09a6097db22a25e : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"21149ec3f92d1611d01694718ec8d8765a8b6d88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 20:16:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c0d5e481951de3f4513d1783e02ec6a5c846511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 22:36:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9749c61a9644c528282b5103d443c3f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dbe95a478b748b6b11b0c69af3cd7c8 : SUCCESS in 11m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e95a985d2b1a4dcfa02bd0eab32759b8 : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/088fab1cbf3d4f2497d59a48db3aa93d : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e454315ae2a041069cc316060fdf8b4a : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b9ef55f72714ac5811621b7c70c98e4 : SUCCESS in 55m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/80e3681588a14796bbfb6b4d52e837b6 : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70654a5f5d2349cd91eab599a0d58c71 : SUCCESS in 57m 00s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e750491d92f54f87b11ba911fa75f904 : SUCCESS in 2h 12m 19s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09bf7e173f4f49b6b7b407dab750ecc1 : SUCCESS in 1h 30m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9976b09db2d14fec9644cd7f4275b1c4 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1d55e66b5c70791183666db497f7291b7ec5e976","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 22:36:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2c3d7363c984b77fdf96d542e4fc12298785decb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 22:37:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d6c99611e8b4c43ad7e876c3d325304\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9435956f7732461f858eb05412920eb3 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9b06052cba70b65d797112c441d244f7efa48aa","revisions":{"b9b06052cba70b65d797112c441d244f7efa48aa":{"kind":"REWORK","_number":1,"created":"2023-11-08 09:01:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/01/900401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/01/900401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/01/900401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/01/900401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/01/900401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/01/900401/1"}}},"commit":{"parents":[{"commit":"79ccde30b45aa3385f253a5fa7057344d2a04fbb","subject":"Split lines to not exceed 160 characters limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/79ccde30b45aa3385f253a5fa7057344d2a04fbb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-08 09:00:57.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-11-08 09:00:57.000000000","tz":60},"subject":"Fix example playbook linters","message":"Fix example playbook linters\n\nChange-Id: If0873fd38d40cb2317e644e20f5be5334753d79a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b9b06052cba70b65d797112c441d244f7efa48aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b9b06052cba70b65d797112c441d244f7efa48aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
