)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~909104","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I6853af8adc53d670ef1935dedebab75c1703a41c","project":"openstack/openstack-ansible-os_cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-14 14:26:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6853af8adc53d670ef1935dedebab75c1703a41c","subject":"Fix rootwrap files idempotency","status":"ABANDONED","created":"2024-02-15 11:02:20.000000000","updated":"2024-05-14 14:26:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"59c9cad40ef6ba792cf821faa3a601fe29741e6d","_number":909104,"virtual_id_number":909104,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-15 13:38:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae51bfe21bf404ad96a500595116a5ceea7b650a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-15 11:02:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5df0ec68c3bc7691dc1ac4c1116168efccb1a822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 13:38:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8155145caeb0491cbf962cf3289a4284\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55812d2a7c834bb68807e831c1462659 : SUCCESS in 13m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/91bebeeae5f6464d8de5b1982ccef1cb : SUCCESS in 1h 12m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/93ded7d4f9454d0e874b2db00383647d : SUCCESS in 1h 14m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5238a51ab0c46668eb76cddc1c85372 : SUCCESS in 1h 21m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5a681ffcd342401784d755ba79cdd5f9 : SUCCESS in 1h 30m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/078ea6fbff4a463aa89400a4a368c6ef : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8759cacf0a124a068a58506bebe63c69 : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82d4fb98ab0a4d50a021a43a12eb38b7 : SUCCESS in 1h 32m 37s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acf4796bf26149548110e581c11b2a4d : SUCCESS in 1h 18m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2b5a9a778af4dd595aa413788e5beaf : SUCCESS in 2h 00m 54s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fc18586b96e140f8a843891b1b7d4699 : SUCCESS in 2h 29m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/077aa925262f4f66964f60b27cf51de5 : SUCCESS in 1h 31m 50s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8cabb8f13bb403b9784be4d6f131b7c : SUCCESS in 2h 20m 49s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/80e23484c4ed4fb7822823b3d9eda341 : SUCCESS in 1h 54m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0026e00e603c41769fc146de944dd85c : FAILURE in 2h 05m 19s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a37f78c580514e44bfd065143b1535a9 : SUCCESS in 1h 04m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ab695207b154587a7e084ce24e1ae1e : SUCCESS in 1h 08m 09s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19ed70a535f84fc8a4e8553b80efb6b6 : SUCCESS in 2h 15m 02s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81a875ce0848482e926204913646427f : SUCCESS in 2h 06m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b937a21169e45e99100e4aa26ce1cde : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ef8b611790a5a83467fb770c87d3f3875591991e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-14 14:26:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"59c9cad40ef6ba792cf821faa3a601fe29741e6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-14 14:26:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c47b98a9bb963e0cf1ae912e6e6b5e419b5535e","revisions":{"43a49934189525d7176046582012b13dca3ab848":{"kind":"REWORK","_number":1,"created":"2024-02-15 11:02:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/909104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/04/909104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/04/909104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/04/909104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/04/909104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/04/909104/1"}}},"commit":{"parents":[{"commit":"7e2cf5283f9cfb3a8a02263a3c0b0bd4d0518a58","subject":"Restart cinder-purge-deleted service only on abnormal exit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7e2cf5283f9cfb3a8a02263a3c0b0bd4d0518a58"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-02-15 11:02:12.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-02-15 11:02:12.000000000","tz":60},"subject":"Fix rootwrap files idempotency","message":"Fix rootwrap files idempotency\n\nTask for ensuring rootwrap permissions adds execution bit to all files\nin folders. Otherwise, rootwrap.d directory won\u0027t have it.\n\nSo using file module with recurse is not giving expected behaviour in any\nscenario.\nWe just remove this task, since expected permissions must be set by prior\ntasks.\n\nChange-Id: I6853af8adc53d670ef1935dedebab75c1703a41c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/43a49934189525d7176046582012b13dca3ab848"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/43a49934189525d7176046582012b13dca3ab848"}]},"branch":"refs/heads/master"},"6c47b98a9bb963e0cf1ae912e6e6b5e419b5535e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-14 14:26:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/909104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/04/909104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/04/909104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/04/909104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/04/909104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/04/909104/2"}}},"commit":{"parents":[{"commit":"62f3ce0efc7cbde90e192dff2199f5c1f21787ae","subject":"Implement variables to address oslo.messaging improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/62f3ce0efc7cbde90e192dff2199f5c1f21787ae"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-02-15 11:02:12.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-14 14:26:25.000000000","tz":0},"subject":"Fix rootwrap files idempotency","message":"Fix rootwrap files idempotency\n\nTask for ensuring rootwrap permissions adds execution bit to all files\nin folders. Otherwise, rootwrap.d directory won\u0027t have it.\n\nSo using file module with recurse is not giving expected behaviour in any\nscenario.\nWe just remove this task, since expected permissions must be set by prior\ntasks.\n\nChange-Id: I6853af8adc53d670ef1935dedebab75c1703a41c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6c47b98a9bb963e0cf1ae912e6e6b5e419b5535e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6c47b98a9bb963e0cf1ae912e6e6b5e419b5535e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
