)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~914086","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I69fee673ebc1fa454a4af0f9a251e9f50999edd8","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-10 08:25:20.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I69fee673ebc1fa454a4af0f9a251e9f50999edd8","subject":"Add service policies defenition","status":"MERGED","created":"2024-03-25 13:08:38.000000000","updated":"2024-05-10 14:13:37.000000000","submitted":"2024-05-10 14:12:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914086-osa/messaging_improvements","meta_rev_id":"8229b9ec262454d4d123b9afb48c864d93bda591","_number":914086,"virtual_id_number":914086,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-10 14:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-07 13:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-08 06:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-08 06:57:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-25 15:40:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 13:05:38.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-08 06:57:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"6334fccbdc50d7e627edceb6e4a9fbe14bce79e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-25 13:08:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d7de32ec1ba8d2c947c286153b801cec9c6c79c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 15:40:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d8d035de7534820a68470da5d2cf3ed\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c40c1cb724d4d099bfefe2469d9b3f2 : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ac97dfdd8a9745dab292c83dc756267a : SUCCESS in 1h 03m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3eb8c7431f5f4812b95226b0ea610eaa : SUCCESS in 1h 29m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5887c9bbb3324d10849e1fc6046a8784 : SUCCESS in 1h 23m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7825e9e317c94e979b21e65b5734e169 : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/049cb68017004181be05f29c747e8ce3 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/468c323ad4ed40aeb9224035cebb85aa : SUCCESS in 1h 33m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cce6370bbef340a5b8ec7d37621bac29 : SUCCESS in 1h 24m 26s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e325146dc6349a1aecec2551410c0af : SUCCESS in 1h 22m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67502e307b924ad88c5ce39c3b4251c0 : SUCCESS in 1h 42m 25s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8246395845864c579cc576538c08269e : SUCCESS in 2h 23m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44f4a5fc47b04c67ae729c10118a9d9a : SUCCESS in 1h 18m 31s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8477ebd34dd45b794bba20e2c985b08 : SUCCESS in 1h 31m 57s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e51ce07700134b2182887de4beb1062f : SUCCESS in 1h 59m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fad25bbb2ad49d185fa9f307ec2756d : SUCCESS in 1h 51m 25s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/72f7cf349b854e559657a7429953f918 : SUCCESS in 1h 04m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bd718c7f6f14ce1b174a344e46a5efb : SUCCESS in 57m 04s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce2cb91d6a944512bf7d2acd72535aa5 : SUCCESS in 1h 56m 33s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0fd23c3e8dc45b6bdbc75abeebd2e0c : SUCCESS in 2h 08m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/691cde7a5f544ed883b61345580178e0 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6c9e199cd957ecacf2c0b735ddd35be668e18cfa","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-07 13:05:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e0bd23a9f05c87bb4256d60c19191d0159662908","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 06:57:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b46399936fbc2f8d69aca6ed7d130bedae522c50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 07:00:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4c4fc6e652a6c0da015180686a9358bf633af6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 08:49:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64ab0d5a60ec45de81973eee7f7e6ecc\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a98b5dff3cf4d3fbf5fccc312514f5b : SUCCESS in 12m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/132850d410ee4a079cce79c824acc153 : SUCCESS in 1h 07m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ee3b2350913d47b9ae0399abe7369d29 : SUCCESS in 1h 11m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/babcbde2052d4616a5fafed4a05eb1a7 : SUCCESS in 1h 25m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c257268242e4d0188e26a2537ca9cbb : SUCCESS in 1h 22m 24s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5433ad96cea4d72939f4c4855bc2ef4 : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/096d83fbd0de47bbb26b67bca7747fda : SUCCESS in 1h 37m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8883cd48a52406c8977e9d100c3ea15 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f88be7f46dd64538ae03c07024e7f165 : FAILURE in 50m 34s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba41536659f14fdf8cd9c37c8c9646a8 : SUCCESS in 1h 41m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5675341cccef467f9c113431bbc45fd1 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0202b97b4680f5f3ceec07f205038dbb085d7eed","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 11:15:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10e7014fc09ed14e6e43d0e749065dcdec8f24da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 13:59:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba89870c312e49439e1a6e2eb2e50a8a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f474e66af79f4b64996936792f2b311c : SUCCESS in 14m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/10fe9fcfd8444038a45cc73840d7f685 : SUCCESS in 1h 16m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c56a929437054165b6f30d8c9af8aa57 : SUCCESS in 1h 27m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb0fd8f88884454c97e315c9e858057e : FAILURE in 1h 00m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3a30bbdb8ea4661a6e1ba07106cdf56 : SUCCESS in 1h 30m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d3dd982c3ec4dc7bc7d6fe27362bc43 : SUCCESS in 1h 21m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/816653cda6324de0b1f5b6b198742c31 : SUCCESS in 1h 24m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9fda6680f6042b68f8e5eec51fe79dd : SUCCESS in 1h 13m 08s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e4a1e8a14a941cc9678b12cd8d9ebd8 : SUCCESS in 1h 28m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e4c4b9b0a1144cec9585acaf11043f15 : SUCCESS in 1h 50m 06s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/830e3eeedd1f4d33885a156969f7288b : SUCCESS in 2h 10m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb8fc715b88949319c91aa2042212a9c : SUCCESS in 1h 36m 00s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de47ba6ff3aa43288a5aa18fae71b0e6 : SUCCESS in 2h 25m 28s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c16f7c5bfecb4cc2ad50467574510a14 : SUCCESS in 2h 07m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c9de89b7b5b4d3b94c06a0d4eb3c2bf : FAILURE in 2h 02m 46s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3a3134ff7d74eccaa07cf93f1dedcf4 : FAILURE in 24m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b44309d02bfd458a94df12e289a158a3 : SUCCESS in 1h 12m 10s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/916cf489899f419c93cbd803ea816fd3 : SUCCESS in 2h 10m 58s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ef42c095fdf4798b638bc4be5788938 : SUCCESS in 1h 59m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f624f370adf465ca96555a3d8f6e9b7 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d8d3c12a7d6f6938d21629cea0e93b77a1b08f30","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-09 07:13:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e58c28f183795ec490bc559657c524edf5edc9a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 10:13:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1fa2612bf844c9d907485f8c5a253da\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0c09d5665404429978508066149111a : SUCCESS in 14m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/47f57ef687874d4b8224885f60e0dcc8 : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/532af260781049f888da34d3b1645622 : SUCCESS in 1h 20m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6153dccd0b54c01884bd9335cbb8873 : SUCCESS in 1h 20m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/afb84f9461e5443595bdbb7c186f7312 : SUCCESS in 1h 08m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a69c41364a444cb9a0c2187c1d749b08 : SUCCESS in 1h 13m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e7c74f8ec1e4c41b051b12251d42f11 : SUCCESS in 1h 36m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/403352bd464a43fdab8967df9f4229b5 : SUCCESS in 1h 24m 19s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef5d81ae0d0744b6976eae972fb4a029 : SUCCESS in 1h 19m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac6d33c5514e402287876818986f5fec : SUCCESS in 2h 03m 46s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de4d692d19b9454da446d5f04b49c01b : SUCCESS in 2h 21m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d60e1c845634ea7a5ff61959cbb89dc : SUCCESS in 1h 58m 31s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2d6e4ca06e748ac969b8c8584d689dd : SUCCESS in 2h 01m 55s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/69f4f935a60e4509bf13604613d267b3 : SUCCESS in 2h 09m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35f57838195b4fa89648d777eeb5513d : SUCCESS in 2h 16m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fc9ab49f2df74540940e69b1a6e05418 : FAILURE in 29m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/165c7dba63754e7795ed88987f8fe6fe : SUCCESS in 1h 01m 54s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5e5e8aac8fc4052ab82e2064ba6263b : FAILURE in 1h 37m 54s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77ca235dc3814ea78cbb96e1ff19a260 : SUCCESS in 2h 07m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece443551eb041eab1800780296391ba : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5b719f6984528c3c179c0d3d677831cad87b8d3f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-09 10:45:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc069aa84cf53b22e67e258eb174c32fb8755a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 13:23:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22447c143def4cb3ba476af41639d58f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba747c5fe2994768828683b3c5bd56b7 : SUCCESS in 13m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2bb4bfacce4641dda3b8a8a7b9ff6b9b : SUCCESS in 1h 18m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/383ab0cfc5e241cea9455d14ef628398 : SUCCESS in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c136f9bb30d240608e4d7ff6cefaf1f7 : SUCCESS in 1h 23m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4f368e0260e40f0bc539809f380dc24 : SUCCESS in 1h 25m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34d1ede14a324513b7c4ceab5ad813b0 : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81b7894ce02246f3ac93398a5f8140cf : SUCCESS in 1h 34m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d6a010b7a7d44bebda02171e7becdad : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62603fcfc0be4bfcb49df2373519ae18 : SUCCESS in 1h 07m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8fe2eabdd54344988755c8d65c6080a3 : SUCCESS in 2h 05m 51s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1fe569bcc894482ea756e770ac96fd4d : SUCCESS in 2h 27m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aba57c5d0cfd4eccb2b6e76291268028 : SUCCESS in 1h 33m 24s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b444ba68a40f449087dedde3a2b22e77 : FAILURE in 1h 56m 30s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c24dad943154e0ebdff213949ca6c0b : FAILURE in 48m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dc3dcd995e14d9cb9655f68e06a3e5b : SUCCESS in 1h 56m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4d6ea6ab72ba4b15bbb86ddcf9263d0a : FAILURE in 26m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e44f61d87d44bf58fbf2a90e1444949 : SUCCESS in 1h 05m 40s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97e8d2a7cca94ef5abca797e87d5e4da : SUCCESS in 2h 03m 50s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8368d6a54004437a1e333f76e4ea626 : SUCCESS in 2h 03m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5717d62955ed4c3a8ba4b46d52fd049c : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2838adc5df8f93793de73fbe7598f3e5e0dd767c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 08:25:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a31f40850759554bbda104a0cecef9526bde7b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 11:33:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/379dfd647b9c4da4afb3045bb4be521d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5291786633024dbca7daf2c35ec8e82e : SUCCESS in 17m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4461181b83d543d8803a4c34fedda0f1 : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/856891515e964b4d8a5f69d25bed1fb1 : SUCCESS in 1h 32m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cc162ea549f4605b74a8d58dc52b854 : FAILURE in 1h 04m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32314a606f0d4b6d8eaae11a5cf5127b : SUCCESS in 1h 42m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd263b74aea843b6a44528d67338cd19 : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dacb6a922a2445b4851bb2d926deede0 : SUCCESS in 1h 35m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c1f3344db694f14ab4a89ef886dda11 : SUCCESS in 1h 28m 30s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b33ab5cb79ee40aca21af5dcca3e96c5 : SUCCESS in 1h 27m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1450fd7455a8426782bef7fe9341b1e1 : SUCCESS in 1h 57m 53s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/437bc6bd47514c8dac9f729367ad4232 : SUCCESS in 3h 01m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15cb6b04254d46b88510b555acd554ca : SUCCESS in 1h 42m 54s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/755fd5d2faac437186c19afbcae251a5 : SUCCESS in 1h 54m 19s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb40a09adc11412cb120614d81c341ff : SUCCESS in 2h 13m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0502c7088a394c61b6ce9f2f245c6ecb : SUCCESS in 1h 53m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f12b1ea5f98f4a93a52211743d9c1dc2 : FAILURE in 26m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/423f3ea83f9d49f28d4b94ac3a72543d : SUCCESS in 1h 10m 23s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87203a3476ed4470b978ba4c28d7461e : SUCCESS in 2h 13m 12s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e939fe09ca0949ec815dfbdc465f6a7c : SUCCESS in 1h 58m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e7abe64b3542c389d7df816f392095 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"29edec68f74e0e587c1fff20693084dc111eccb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 11:36:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65484c695486963208d51fcebcdc95fe56d151a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 14:12:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d56852c1ee8b4f1d9a521a2ad19a2bb0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/077d97e1f3f947d98527a6d68988ceea : SUCCESS in 13m 27s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5efac00f0200490eb6cbbd283ace1ad3 : SUCCESS in 1h 39m 08s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/83dd50b5826641f89799c1c2acc454fa : SUCCESS in 1h 28m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2403008feadf4acdb96d984c2a280ef3 : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f72e09f6de740baabfafd6e6108846c : SUCCESS in 1h 36m 50s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/becc3e08580944e48f1af9be6cc756f8 : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/229a71aeee1949849614f02540e0d386 : SUCCESS in 1h 39m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dad600a28284920bc4d9b0b58d9f84b : SUCCESS in 1h 25m 25s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffa059a9a44d4279a326382d52731525 : SUCCESS in 2h 32m 51s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1d55a1a2f7d4987ac8491080d92de7e : SUCCESS in 2h 08m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/795ad3c61db344bb847ab295888ab17b : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"292efdac54d156a350edca5979082eb5b3dadfd2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 14:12:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8229b9ec262454d4d123b9afb48c864d93bda591","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 14:13:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faaa63d2c6af48f3ae50f884ac51275d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c413c6ae126a458d90026bca96886344 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ad37dd6209c12259b0efe2204a20ee4edc82f4e","revisions":{"5ad37dd6209c12259b0efe2204a20ee4edc82f4e":{"kind":"REWORK","_number":1,"created":"2024-03-25 13:08:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/914086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/86/914086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/86/914086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/86/914086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/86/914086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/86/914086/1"}}},"commit":{"parents":[{"commit":"7e2cf5283f9cfb3a8a02263a3c0b0bd4d0518a58","subject":"Restart cinder-purge-deleted service only on abnormal exit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7e2cf5283f9cfb3a8a02263a3c0b0bd4d0518a58"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 13:08:20.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 13:08:20.000000000","tz":60},"subject":"Add service policies defenition","message":"Add service policies defenition\n\nIn order to allow definition of policies per service, we need to add variables\nso service roles, that will be passed to openstack.osa.mq_setup.\n\nCurrently this can be handled by leveraging group_vars and overriding `oslomsg_rpc_policies` as a whole, but it\u0027s not obvious and\ncan be non-trivial for some groups which are co-locating multiple services\nor in case of metal deployments.\n\nChange-Id: I69fee673ebc1fa454a4af0f9a251e9f50999edd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/5ad37dd6209c12259b0efe2204a20ee4edc82f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/5ad37dd6209c12259b0efe2204a20ee4edc82f4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
