)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~914100","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~Ifb16270b41eb795dca9b4bd6a1a3e6eabf2f45eb","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-10 15:05:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ifb16270b41eb795dca9b4bd6a1a3e6eabf2f45eb","subject":"Add variable to globally control notifications enablement","status":"MERGED","created":"2024-03-25 15:18:58.000000000","updated":"2024-05-10 20:40:41.000000000","submitted":"2024-05-10 20:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914100-osa/messaging_improvements","meta_rev_id":"fd866473e5b9d0a9088467df54bc248e0c1c3a2f","_number":914100,"virtual_id_number":914100,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-10 20:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-07 13:05:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-08 06:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-08 06:57:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-25 17:44:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 13:05:55.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-08 06:57:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f55b63ca11a9e5636905d1df04dc98a9eaa1bb1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-25 15:18:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa35a8a04f30a88df9ba609b504fd49330883699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 17:44:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552336c0ce964eed99e9aa99ff9baacf\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/017cd2fe66f94e5f9878622dd8943af5 : SUCCESS in 10m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b478913c888c4ad9bea33f17704913ea : SUCCESS in 1h 08m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ac49f2c39e1429cbd76db6c23c8f7e3 : SUCCESS in 1h 18m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f86ff95a7ab84405a1174e9c60dc3b2f : SUCCESS in 1h 18m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d1aa75e2d72b48f49b91d5496c8db12c : SUCCESS in 1h 27m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/339ccc5327aa4731a298bc13f98fc126 : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/893370cd0cd7436fa10b8ce01bb6193b : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a2454017ed84f25aec459c0416fd2b5 : SUCCESS in 1h 26m 00s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ad8e491b7a742939b0b4c656e28914c : SUCCESS in 1h 08m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/70473b7f640b42288d842f65ae689b64 : SUCCESS in 2h 00m 41s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59e7ed85ee854882bd5bfb02b4fb7f22 : SUCCESS in 2h 18m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c26403c4e4044599ea75e15ebb0a709 : SUCCESS in 1h 26m 07s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47708cf4d2b9467cb3a9f2ce01e2d356 : SUCCESS in 2h 08m 17s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6062cc7839284f4c87934582a39636d7 : SUCCESS in 2h 04m 08s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29e3293fe2ba4dceaba7d92b71334803 : SUCCESS in 2h 01m 44s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fead0be3cd474261965cbecbabcbe032 : SUCCESS in 1h 06m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2daaea7184224650bedadc2f3ed58b16 : SUCCESS in 1h 02m 31s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e7a0bbb04b54fbdb566ff0fcd69a96c : SUCCESS in 2h 16m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f6470b694b34f0880db9b041ffe369e : SUCCESS in 2h 03m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f8ac8d1ec645aab7e5cfda5e54b9bc : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fb6054c35ae10518b947588c35af07c0f97b8460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-25 18:03:43.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0d6db3e885d1ebd845eafc4c44856e1c1fea019e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-25 18:15:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a5d26b10a4a34b05ce0366fdec0dc41aef8de9f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-25 18:23:55.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"605ed6242c99e95d6daed93d6d893e67d6ad6b03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 20:47:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170242d9ba5a4f339a4fb345907e95b2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fd246b8616d4cb4823ce893db775027 : SUCCESS in 11m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d890c00c059f4009b666eb8fd9c78985 : SUCCESS in 1h 15m 50s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/93c73c65f17d4c269957868c4d93829a : SUCCESS in 1h 21m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d591622f25445d590c796f0f8221e01 : SUCCESS in 1h 29m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/613d91d1948f483792a2657fe8eb0ef2 : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4188c7c914845e1b65fd043578b47b2 : SUCCESS in 1h 28m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/80db8cb9d4b1408f9cf94e514f43fb81 : SUCCESS in 1h 33m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba9731ea15274e06bc8fa666492c7c93 : SUCCESS in 1h 12m 14s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e928bbc5eab94519a089b6f30f3c729a : SUCCESS in 1h 23m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f5e5764a6ae147448b76cf19b0b14ee8 : SUCCESS in 1h 49m 26s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/96b715789a2b445981d6afd3080344ba : SUCCESS in 2h 18m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8b28c4b3a794c359eea7f8606649c34 : SUCCESS in 1h 33m 05s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0c30e18655b49db9cc8f8c9f9fff959 : SUCCESS in 2h 15m 51s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d130efd8347841ffabde8c124b4dbe26 : SUCCESS in 2h 08m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40d9cfd6989347fdb6b7557ac172ebc9 : SUCCESS in 1h 36m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4bc82c08960c45eaa32a5ff1c0c0a6ec : SUCCESS in 57m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a974691efa343dca9ec9bc1300ede97 : SUCCESS in 1h 06m 17s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/144d4b649e614717a66b8a5c2267108d : SUCCESS in 2h 20m 59s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62c1c1656cfe4d9ab70d348d06d84010 : SUCCESS in 1h 52m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6effc1ebffd74468b478ae25f1d068b3 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":4},{"id":"d0cbce29a43ee1a1002247a2d82e343e737be63d","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-07 13:05:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"12805dac427bd7843436a35fa2d655cfd38791e6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 06:57:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"dd869ce1a74cd7bb5f6a9c8787b8476655267b69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 06:57:40.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63bd8ffe0d514613b7a42a4a27d86e66\n\n\nWarning:\n  Change 914144 in project openstack/openstack-ansible-plugins does not share a change queue with 914100 in project openstack/openstack-ansible-os_cinder","accounts_in_message":[],"_revision_number":4},{"id":"e0cb58e387b77f74005c5e701dbd7f34a27ef7d5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-09 06:34:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e03787c5b34a6e5ab42dffe366fdddb8e8f5a3d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 09:14:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c7c2c686b984241a47f5716fe21d2fe\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a87c190804ab4512bf923193b536e5dc : SUCCESS in 14m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5ba24cd72e0c4656a226faac5a38189c : SUCCESS in 1h 23m 31s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/11cccdc170a7486fbaa6e7bac9b70822 : SUCCESS in 1h 09m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d144baff61c4ab7962fd3b7d722b9a5 : SUCCESS in 1h 34m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c2828bdedbfd49648af5ade8a6ac8769 : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9c4017ac40446a49673f9a895cfc831 : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/730a9243f9e84b81b7fb391d7befa076 : SUCCESS in 1h 50m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9d3dd044f6841aab493e5d0249dd731 : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58cd002469c54089bdd1f58970c8fa53 : SUCCESS in 1h 30m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7238b3a8e3be4583bf3840ba8b2edd1e : SUCCESS in 1h 46m 50s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b07054a78e9147468c9152b6205695cb : SUCCESS in 2h 21m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/544948d9d2a24ed6926b70c331a4f626 : SUCCESS in 1h 28m 53s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e03cc5db21d4267ba93bdd00630fd4e : SUCCESS in 1h 47m 14s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2b30f687c9341c68f78bfffed4d0c3b : SUCCESS in 2h 19m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee8aa23a5f9543f4865198d4412ec91a : SUCCESS in 1h 53m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11db4706343447fa99986b11b8396c8f : FAILURE in 27m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7ab81d980ff426aa7af79b10aa6b1e2 : SUCCESS in 58m 13s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c8a135459984df5a91fcc10bacc065e : SUCCESS in 2h 19m 27s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a74703cf71e341d78a45eb4ce7a93338 : SUCCESS in 1h 52m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad09a472eafc41fb80b0ad9f410bb2d7 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":4},{"id":"aa53c9b0f8134d0c38bd0331a6de8d2f9254c08b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 11:36:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e983c928aa3709fec1da44328bfbe70e8f89cebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 14:37:16.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a2705f17fa646ecabe97f393a9a6cb9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adba8e940cfe469ba9fc2ce48405e731 : SUCCESS in 13m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dde2c96157b34314ad2d869b6afcb469 : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0f4296da608d41bd8cd1272856cb2c33 : SUCCESS in 1h 31m 17s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c9aa91149e8b41dc80edbd4b89d2c149 : SUCCESS in 1h 33m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f806b691f0e4878a288bc9437723a64 : FAILURE in 1h 17m 20s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ba5c3160f6245c588283581279bd400 : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e024bfcf7c84b3f882bf11da671aa86 : SUCCESS in 1h 30m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/970f25bc3a214beb8841544c76607821 : SUCCESS in 1h 26m 51s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40cfd3ce4a0b4f0a9fd912815b477aeb : SUCCESS in 2h 32m 46s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbbf3fa146704a57bd78e18720fe6cd6 : SUCCESS in 2h 55m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312fd0c0c183435ba07ecf06982fbcdf : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":4},{"id":"c0cbb4e61d7e35ed181c466e83ab7b25c9296460","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 15:05:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"56629d14fd0aa961360237ca3d82274a74cf9054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 17:44:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47bfc914c8ab40c1a7512f8890251ab1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0f615b1f7c04a2aada5592134409977 : SUCCESS in 14m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/480fc086971340c9a073550744bc70d4 : SUCCESS in 1h 23m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/55c31496c4cd477cafe69deee8b0c810 : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cacf22fa6a1142f5ae0b83a350cb4199 : SUCCESS in 1h 25m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1630a36986414b73ba6ce0c7ebc2f8f7 : SUCCESS in 1h 25m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12f64b1e1c024c9db2c7b03f607f6366 : SUCCESS in 1h 24m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/754f3d69af0147218796753552f4ce2c : SUCCESS in 1h 28m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d6e353d130c4d4e9a0b9b5da213d335 : SUCCESS in 1h 24m 55s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a1611f0f485458c986f444988119b00 : SUCCESS in 1h 19m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83eecf8df1bf474cab333d8ee00b85dc : SUCCESS in 1h 47m 54s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/75e01faf7ff64ec897d7f3f34e3def10 : SUCCESS in 1h 59m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dc74d5316404def97b901608a9ec55a : SUCCESS in 1h 52m 27s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c8c37e7337744a9a18de514da8180fa : SUCCESS in 2h 13m 48s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/92cd8484adfe4fe7ae0399b45a45e26d : SUCCESS in 2h 21m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd9326a8a0cd4170975eb0ec48ebfbfd : SUCCESS in 2h 01m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f01863879b1455981aada3524f7c789 : FAILURE in 28m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8892d5cfdd0d4f11b071bcd873e314bb : SUCCESS in 1h 06m 18s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8251739e7f95408098bae370bc61b4fa : SUCCESS in 2h 28m 06s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ac6b14276434ee2a0a44233456c07a1 : SUCCESS in 2h 14m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed8843db58445d997a916849ab7e407 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4},{"id":"44285c2dced4e79a3434a400415ce6e7f1518a62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 17:47:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"568d84154f8a0140f339488278544f559db0e9d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:39:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e35717f0b704c809f0cb0fd35153f23\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52a30b4cc9f04a73983c3440a9ffd084 : SUCCESS in 12m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1b6ec0ed3f414f15b1472a051315485c : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/40d7b694bed74ed49846acc16950d065 : SUCCESS in 1h 04m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83a19e69e5e54d25963d4e544a095c31 : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2729eb53f354773b44a74864c0f8a3e : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73930ecd29b24b96a7cf4d119eac0993 : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91cb9a06241342bfb86ca0c1f30dac07 : SUCCESS in 1h 42m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/494c688a6ace42d5a91368f42a08799d : SUCCESS in 1h 15m 09s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/300fa5bb00804a6d8a1842b793463c5e : SUCCESS in 2h 42m 00s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4efe5c75e6754a1a99fe0c79515d9cd3 : SUCCESS in 2h 09m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276fe3916ddb4c50973679616f125753 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"f293eb433a42584807a362d183cc52794500db5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:39:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fd866473e5b9d0a9088467df54bc248e0c1c3a2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:40:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85b070d31c274adaae98ca154f4831a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c7b29650bd4796bde39c84f4bd876a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"21f8ede972afa3b5fb4ea6055a79daa78b8e0bf7","revisions":{"cda8627d1a7a9cc8cdc1365df0f18ad21f4a840b":{"kind":"REWORK","_number":1,"created":"2024-03-25 15:18:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/914100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/00/914100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/1"}}},"commit":{"parents":[{"commit":"5ad37dd6209c12259b0efe2204a20ee4edc82f4e","subject":"Add service policies defenition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/5ad37dd6209c12259b0efe2204a20ee4edc82f4e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 15:18:48.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 15:18:48.000000000","tz":60},"subject":"Add variable to globally control notifications enablement","message":"Add variable to globally control notifications enablement\n\nIn order to be able to globally enable notification reporting for all services,\nwithout an need to have ceilometer deployed or bunch of overrides for each\nservice, we add `oslomsg_notify_enabled` variable that aims to control\nbehaviour of enabled notifications.\n\nPresence of ceilometer is still respected by default and being referenced.\n\nPotential usecase are various billing panels that do rely on notifications\nbut do not require presence of Ceilometer.\n\nChange-Id: Ifb16270b41eb795dca9b4bd6a1a3e6eabf2f45eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/cda8627d1a7a9cc8cdc1365df0f18ad21f4a840b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/cda8627d1a7a9cc8cdc1365df0f18ad21f4a840b"}]},"branch":"refs/heads/master"},"51ef5ba657b17f7fec7ca89dc61a6b323b80da92":{"kind":"NO_CHANGE","_number":2,"created":"2024-03-25 18:03:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/914100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/00/914100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/2"}}},"commit":{"parents":[{"commit":"5ad37dd6209c12259b0efe2204a20ee4edc82f4e","subject":"Add service policies defenition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/5ad37dd6209c12259b0efe2204a20ee4edc82f4e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 15:18:48.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 15:21:44.000000000","tz":60},"subject":"Add variable to globally control notifications enablement","message":"Add variable to globally control notifications enablement\n\nIn order to be able to globally enable notification reporting for all services,\nwithout an need to have ceilometer deployed or bunch of overrides for each\nservice, we add `oslomsg_notify_enabled` variable that aims to control\nbehaviour of enabled notifications.\n\nPresence of ceilometer is still respected by default and being referenced.\n\nPotential usecase are various billing panels that do rely on notifications\nbut do not require presence of Ceilometer.\n\nChange-Id: Ifb16270b41eb795dca9b4bd6a1a3e6eabf2f45eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/51ef5ba657b17f7fec7ca89dc61a6b323b80da92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/51ef5ba657b17f7fec7ca89dc61a6b323b80da92"}]},"branch":"refs/heads/master"},"cd6452a83c268a30ca7ffd94ec269b1533571027":{"kind":"REWORK","_number":3,"created":"2024-03-25 18:15:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/914100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/00/914100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/3"}}},"commit":{"parents":[{"commit":"5ad37dd6209c12259b0efe2204a20ee4edc82f4e","subject":"Add service policies defenition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/5ad37dd6209c12259b0efe2204a20ee4edc82f4e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 15:18:48.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-25 18:15:04.000000000","tz":0},"subject":"Add variable to globally control notifications enablement","message":"Add variable to globally control notifications enablement\n\nIn order to be able to globally enable notification reporting for all services,\nwithout an need to have ceilometer deployed or bunch of overrides for each\nservice, we add `oslomsg_notify_enabled` variable that aims to control\nbehaviour of enabled notifications.\n\nPresence of ceilometer is still respected by default and being referenced.\n\nPotential usecase are various billing panels that do rely on notifications\nbut do not require presence of Ceilometer.\n\nChange-Id: Ifb16270b41eb795dca9b4bd6a1a3e6eabf2f45eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/cd6452a83c268a30ca7ffd94ec269b1533571027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/cd6452a83c268a30ca7ffd94ec269b1533571027"}]},"branch":"refs/heads/master"},"21f8ede972afa3b5fb4ea6055a79daa78b8e0bf7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-03-25 18:23:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/914100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/00/914100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/914100/4"}}},"commit":{"parents":[{"commit":"5ad37dd6209c12259b0efe2204a20ee4edc82f4e","subject":"Add service policies defenition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/5ad37dd6209c12259b0efe2204a20ee4edc82f4e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 15:18:48.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-25 18:23:55.000000000","tz":0},"subject":"Add variable to globally control notifications enablement","message":"Add variable to globally control notifications enablement\n\nIn order to be able to globally enable notification reporting for all services,\nwithout an need to have ceilometer deployed or bunch of overrides for each\nservice, we add `oslomsg_notify_enabled` variable that aims to control\nbehaviour of enabled notifications.\n\nPresence of ceilometer is still respected by default and being referenced.\n\nPotential usecase are various billing panels that do rely on notifications\nbut do not require presence of Ceilometer.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/914144\nChange-Id: Ifb16270b41eb795dca9b4bd6a1a3e6eabf2f45eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/21f8ede972afa3b5fb4ea6055a79daa78b8e0bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/21f8ede972afa3b5fb4ea6055a79daa78b8e0bf7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
