)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~914143","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I02db2ad91ad036ff24c35882e1746b3d6a8f3cc0","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-05-13 10:45:10.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-13 13:14:36.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-05-13 10:53:09.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I02db2ad91ad036ff24c35882e1746b3d6a8f3cc0","subject":"Implement variables to address oslo.messaging improvements","status":"MERGED","created":"2024-03-25 18:03:43.000000000","updated":"2024-05-13 13:15:37.000000000","submitted":"2024-05-13 13:14:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914143-osa/messaging_improvements","meta_rev_id":"6ca09f776628b884827695f1986a985b7e6bbafe","_number":914143,"virtual_id_number":914143,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-13 13:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-13 10:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-13 10:53:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-13 10:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-25 21:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 13:06:11.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-08 06:57:10.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"8dba532b802e41f9c47ac3cb9a9a680e59aa22bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-25 18:03:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e1c714d0aeec61b973e004de557fceb9edfe6e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-25 18:15:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b22f95547130c746d5c0b514140a2119a9187d58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-25 18:24:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d4615a8f0eb8a021b4f56e05574608960436d229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 21:02:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4b02f1d1251482c8556d46ced067bc1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba04308fb0a146259fa2e1b5e5bcbed0 : SUCCESS in 11m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d3e0732124d34f978d0544ad87fcca64 : SUCCESS in 1h 13m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b71374dfd70e447b8a3ad9e1157c82e2 : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfecae3ce29d4b1787c3c5f9b6d9235c : SUCCESS in 1h 31m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d8cfbc11a6d74bffb945f55fd541c865 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/807f1dbf5ae446c1a23037fa00c1e506 : SUCCESS in 1h 11m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e7de75477bc6460c8f0d8c6a658e9390 : SUCCESS in 1h 27m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef0e51b42e87465c9ccf2d7dad4e1297 : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67d6c43bac114916aad7ff6521f9ee89 : SUCCESS in 1h 18m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4d50dc9fa7944d6f8460f07d01f345b7 : SUCCESS in 2h 04m 18s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be95378561cc4e289a211f4eaaeab627 : SUCCESS in 2h 20m 03s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e750541cb3934767bd68670ce025d1e3 : SUCCESS in 1h 42m 06s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1068962d6ae745cbadcf8ec4b2036ab0 : SUCCESS in 2h 21m 05s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/60ed12ccc4ec4f20a5632a5c9c10b7f2 : SUCCESS in 1h 58m 26s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6057402709f84082aab95872c4ec84e3 : SUCCESS in 2h 17m 13s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f481e2af186f4d8992e9155213d81e77 : SUCCESS in 1h 06m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4646be9434f6435fa65c8d1044e91d34 : SUCCESS in 1h 02m 28s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/897f6c15a1c146e79629454a10702b02 : SUCCESS in 2h 13m 50s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb141237d9c54d8ab4cf0fa47ed27488 : SUCCESS in 2h 06m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da6136fc685043d99610f94e65d0801a : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ca44c0af2b9ede0fb2450d3c1baf0e9007b8d122","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-07 13:06:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c5595178eee4c5b8365475be1380e973ddf36fa5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 06:57:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b832fc0549d113261c1b937ea268e793a755955a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 11:36:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e97c07523fa0c124635906eb53a90f44eefc9264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 14:37:17.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/914100 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"4245adb12765f9cfc45cfd1e083bea3ee3676aa4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-12 09:41:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e9b96eb73a9c4338a7596d950ca6b95026f614ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-12 12:18:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f493711eba84b849d831b1c72cfcf87\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4fd7bb7e99540e1bd126cc6c8a378bf : SUCCESS in 13m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/65f6ff6603d1445a94a0bfd5ff1d5d32 : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8083fb26a5714d59a043b5a825894859 : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/892989c756634f2386548e989b5b48f6 : SUCCESS in 1h 25m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b11d7e44ede44504806759b93fb59fa1 : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7504feec95cd4111a2440fa325fcb9bf : SUCCESS in 1h 16m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a8397d3afff4404483d33f385f911827 : SUCCESS in 1h 33m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a4f502260864b0bb30d713a3c436726 : SUCCESS in 1h 21m 19s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a0f090a790548c68dda40cbce9d5f92 : FAILURE in 1h 12m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b707b9fc457b4def8805191d7435b25e : SUCCESS in 1h 53m 57s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2fdf01a032b24c499faa22ee47bc99af : SUCCESS in 2h 30m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2259968c3bc84d63b74a5b5e2e86f843 : SUCCESS in 1h 24m 28s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a37191b351814e27adcc2b8b1268e1f9 : SUCCESS in 1h 35m 58s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/898958cd95fe48808fa4e5195a1a6bdf : SUCCESS in 2h 03m 56s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2f1452059f84e4584da08e820a92359 : FAILURE in 1h 36m 25s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/da8ec5a407ea45a6a5074052a9618fa5 : FAILURE in 23m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/242ae072386648519ea16f5870d26a49 : SUCCESS in 53m 10s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccc8aa1d4f7a443f9a298a68ef1dca3b : SUCCESS in 1h 51m 52s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3b8333e56bc4e2eabb35b7e85015239 : SUCCESS in 1h 38m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ecd6ab672324595bb32cfe3a00cbbe7 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d405718b543d3228ca6f9162e1930bfcd3df587f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-13 07:21:38.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a07e11a7e70b6d1e201ee2b56807ae733bd46296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 09:50:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10fe8cebe23f465f8dd8395b6d7c3497\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5edfca847e246ce8b00eb9a8d20fa1e : SUCCESS in 14m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bc8e7f4df9844eeea2883773f6838816 : SUCCESS in 1h 02m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0bea344326c04567a6f9e75dcda1bc0a : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fef2ffd96e2a49638d916ba3f4240210 : SUCCESS in 1h 12m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41f1a295a4e94056aa832aed946b958f : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e67f62118f084592b621d41446c29614 : SUCCESS in 1h 29m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73540d66cb504bcbbf22fbce0c78a87d : SUCCESS in 1h 24m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab47de52a1794e299e0da847caa16d18 : SUCCESS in 1h 23m 15s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbd32d606b984af180900fa1aefc24e7 : SUCCESS in 1h 10m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c7e56bd69265443fad24cfc2199d08fd : SUCCESS in 1h 54m 22s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5322fc466a664053b2d2fb10662883a0 : SUCCESS in 2h 06m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e68588ac73f949bbbad6188eaabd67fa : SUCCESS in 1h 56m 39s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70ae0571b8494a5689181f9ad396ffc8 : SUCCESS in 2h 15m 09s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af537ecbe739494aaff313113cac9eb8 : SUCCESS in 2h 03m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d517bc7e13446a2ac0ea6c6a1b6cc2d : SUCCESS in 2h 08m 06s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce48f65b1ee340cd8c7085a976476226 : FAILURE in 27m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43df71031ce44152b3a06ce248721ab9 : SUCCESS in 56m 15s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb08a3afae8045b69a8ac7a72ba0ffec : SUCCESS in 2h 21m 03s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a098f63bf64f4395be20d8ffe035d6ab : SUCCESS in 1h 48m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb65aff5f1c4cee869a045e2dd9d9dd : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"5fa82d5d8ea31ee967b3646bb7028cabd9e8d546","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-13 10:45:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3ab875675df716b765f727719913de35f1351a75","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-13 10:53:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"38efca317604eee0d0c68bfaa4e2028447e7995e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 10:55:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1eef23bc80636879573dd8164b4e4b7e636891c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 13:14:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3654d76333a6428c896bc21ad99d8649\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a2aa54042774ffba2579e4376844962 : SUCCESS in 13m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/993bf1b2e9054d0b9ae366155defbe1f : SUCCESS in 1h 09m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8bb796192a1147998fb8c9c0efd26ca0 : SUCCESS in 1h 24m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63441c5b76f346d5897802aff28bd30f : SUCCESS in 1h 33m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e16fd1cbe34344a1ba292c336142bd50 : SUCCESS in 1h 25m 11s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85e466ef2fa04608b8ad45f83b5be013 : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0b97068e5b24af2a27c3125a58054fa : SUCCESS in 1h 36m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9143e7fd6e91430c83c4394647f01a7b : SUCCESS in 1h 28m 20s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99f6a612b6394c869d8e2ec46c36fd80 : SUCCESS in 2h 08m 24s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0087115ce08347b8a80eba71a0116bc2 : SUCCESS in 2h 01m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb82e31f97f4cdb8ef7b41670eacc9e : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"0b2f5c69497672d18edb1e81763e1741e9410c92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 13:14:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6ca09f776628b884827695f1986a985b7e6bbafe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 13:15:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e38b27e7b6bf41a48b4ac55e15a0994f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665c0fc20d3d48b4bc0fd9953b71334e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"62f3ce0efc7cbde90e192dff2199f5c1f21787ae","revisions":{"92a4211604928603f9320e671c85333eecfda6af":{"kind":"REWORK","_number":1,"created":"2024-03-25 18:03:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/914143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/43/914143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/1"}}},"commit":{"parents":[{"commit":"51ef5ba657b17f7fec7ca89dc61a6b323b80da92","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/51ef5ba657b17f7fec7ca89dc61a6b323b80da92"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 17:51:48.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 17:51:48.000000000","tz":60},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: I02db2ad91ad036ff24c35882e1746b3d6a8f3cc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/92a4211604928603f9320e671c85333eecfda6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/92a4211604928603f9320e671c85333eecfda6af"}]},"branch":"refs/heads/master"},"6a95be2a7f23c1f7d2386581c57e5ce690b2f98e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-25 18:15:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/914143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/43/914143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/2"}}},"commit":{"parents":[{"commit":"cd6452a83c268a30ca7ffd94ec269b1533571027","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/cd6452a83c268a30ca7ffd94ec269b1533571027"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 17:51:48.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-25 18:15:53.000000000","tz":0},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: I02db2ad91ad036ff24c35882e1746b3d6a8f3cc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6a95be2a7f23c1f7d2386581c57e5ce690b2f98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6a95be2a7f23c1f7d2386581c57e5ce690b2f98e"}]},"branch":"refs/heads/master","description":"Rebase"},"a70dc096db16dc4a1520b3488815b17d0d076dd6":{"kind":"NO_CHANGE","_number":3,"created":"2024-03-25 18:24:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/914143/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/43/914143/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/3"}}},"commit":{"parents":[{"commit":"21f8ede972afa3b5fb4ea6055a79daa78b8e0bf7","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/21f8ede972afa3b5fb4ea6055a79daa78b8e0bf7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 17:51:48.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-25 18:24:00.000000000","tz":0},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: I02db2ad91ad036ff24c35882e1746b3d6a8f3cc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a70dc096db16dc4a1520b3488815b17d0d076dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a70dc096db16dc4a1520b3488815b17d0d076dd6"}]},"branch":"refs/heads/master","description":"Rebase"},"62f3ce0efc7cbde90e192dff2199f5c1f21787ae":{"kind":"REWORK","_number":4,"created":"2024-05-13 07:21:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/43/914143/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/43/914143/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/43/914143/4"}}},"commit":{"parents":[{"commit":"21f8ede972afa3b5fb4ea6055a79daa78b8e0bf7","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/21f8ede972afa3b5fb4ea6055a79daa78b8e0bf7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-25 17:51:48.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-13 07:21:38.000000000","tz":0},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: I02db2ad91ad036ff24c35882e1746b3d6a8f3cc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/62f3ce0efc7cbde90e192dff2199f5c1f21787ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/62f3ce0efc7cbde90e192dff2199f5c1f21787ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
