)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~919083","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~Ic25c44d057a1aea57ad250a0c7cac33712f2017b","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/messaging_improvements","hashtags":[],"change_id":"Ic25c44d057a1aea57ad250a0c7cac33712f2017b","subject":"Add qos_prefetch_count to variables","status":"ABANDONED","created":"2024-05-10 16:38:15.000000000","updated":"2024-05-13 07:21:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b26e8ef6af8460deeeb074710a5fd3c36b9ac4f5","_number":919083,"virtual_id_number":919083,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-10 19:34:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-10 19:34:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"746426759f36afb182cf0a56a22ba4a5e2f3ba03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 16:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21ee3496dd4d8793294ee59fdde7a15ce7731f29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 19:34:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c435997cbd0843c6b786cd86b7ecaf9a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e478bde75b1464a859f217a795ad6b4 : SUCCESS in 13m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ef0803be58904f7fa6ce90e86848da23 : SUCCESS in 1h 20m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bee165be73b24989a91a82033a74c73d : SUCCESS in 1h 21m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5fbf512f9f1496e8c46bdeeee17ac44 : SUCCESS in 1h 23m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f71b0b7e5fc74ac69689da8aad06ac45 : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5f624526f4a490e97205e3a3bc9ec0a : SUCCESS in 1h 14m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c0e8de6fe2dc4b37945e5fc42a5660bc : SUCCESS in 1h 34m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7c66e1378d34e4fb084dc85c6b4ed10 : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48261bd6f8a14dbea8de839b78e2fddd : SUCCESS in 1h 18m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/75c2b4b86a284f81b0a34cbb0e4e2338 : SUCCESS in 1h 55m 28s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6362b39146814564a518628152444482 : SUCCESS in 2h 40m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfa3d58507514f3e9b979ceccfd458ff : SUCCESS in 1h 32m 49s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6862650a1cbb4e8db6763682888a9710 : SUCCESS in 2h 29m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf33696961db4ae2a99a725e9588c4fb : SUCCESS in 2h 09m 09s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70f677a822214a5c927bcaa156116bf1 : SUCCESS in 1h 49m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/240f0303c5ad4255a94ae2bec70d9ade : FAILURE in 24m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/138feb55cb3e4d62ad51665fbeac350f : SUCCESS in 1h 10m 59s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf6f245a77ce46f49bc919441c96499b : SUCCESS in 2h 37m 10s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/951270377ecb4412a6752951e13cb945 : SUCCESS in 2h 05m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c0cfcd83cd475e913ccffed5c88ab6 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b26e8ef6af8460deeeb074710a5fd3c36b9ac4f5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-13 07:21:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73020d903accd59b6c8a6d93d16d1b597173e2b6","revisions":{"73020d903accd59b6c8a6d93d16d1b597173e2b6":{"kind":"REWORK","_number":1,"created":"2024-05-10 16:38:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/919083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/83/919083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/83/919083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/83/919083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/83/919083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/83/919083/1"}}},"commit":{"parents":[{"commit":"a70dc096db16dc4a1520b3488815b17d0d076dd6","subject":"Implement variables to address oslo.messaging improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a70dc096db16dc4a1520b3488815b17d0d076dd6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 16:38:05.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 16:38:05.000000000","tz":120},"subject":"Add qos_prefetch_count to variables","message":"Add qos_prefetch_count to variables\n\nWhen rabbit_stream_fanout is set to true, rabbit_qos_prefetch_count\nshould be set to a positive value.\n\nThis also fixes defenition of transient_quorum_queue as if\nrabbit_stream_fanout is enabled - transient_quorum_queue should be also\nenabled. [1]\n\n[1] https://github.com/openstack/oslo.messaging/blob/master/oslo_messaging/_drivers/impl_rabbit.py#L726-L736\n\nChange-Id: Ic25c44d057a1aea57ad250a0c7cac33712f2017b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/73020d903accd59b6c8a6d93d16d1b597173e2b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/73020d903accd59b6c8a6d93d16d1b597173e2b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
