)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~919674","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I6f4d5f3388b71ce874650558b2566f795361a6c7","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/rmq-migrate","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-05-20 10:34:20.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-20 10:34:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f4d5f3388b71ce874650558b2566f795361a6c7","subject":"Add tag to enable targeting of post-install config elements only","status":"MERGED","created":"2024-05-15 08:16:05.000000000","updated":"2024-05-20 10:35:06.000000000","submitted":"2024-05-20 10:34:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919674-osa/rmq-migrate","meta_rev_id":"2a01f9b9868831c94ca61f201f28692b74eabfca","_number":919674,"virtual_id_number":919674,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-20 10:34:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-05-17 15:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-16 20:41:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-20 08:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 11:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-16 08:32:20.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2024-05-16 20:41:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-05-17 15:53:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"9212affe3ddaec88333cc96dc9c19401c16820c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-15 08:16:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c0b489be2ded4cb42a508f8b0a216e7ac073c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 11:19:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4738892ebfbc4c62b4444349f62594ff\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96099b8434914f5da69f0414ee58fb60 : SUCCESS in 12m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/abb779f32bdd49b8abb7c1bc22781e50 : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/45bf0123680045059c417d3828179a49 : SUCCESS in 1h 21m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c67bf8bcdc174fc8aae38c163b33348e : SUCCESS in 1h 32m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d92d8b6e1a6c48638d344d709c75f6a8 : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2f4dd1c1f4d44a3939a837661b41984 : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/269c3e11af3644d5a6627f7f7c5e43db : SUCCESS in 1h 38m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b74f54dfa4d4458eaf3a099c1976a171 : SUCCESS in 1h 29m 29s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b5cbad1047b4c90a939c30d9f6c592f : FAILURE in 1h 26m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6861f6bd5cf4c51bd3e64c8769bc3fd : SUCCESS in 1h 54m 22s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d21a7093686f4cfca74a6141299fb63b : SUCCESS in 2h 29m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6ed5f62127b41ffaa0b0530350cbe6a : SUCCESS in 1h 57m 57s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/370dfc1ec6134bb4b190eec7e8768cfb : SUCCESS in 2h 07m 41s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3975a0a610e34eb8bdf82e0ff27450dd : SUCCESS in 2h 06m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03c6f7952ff84d5b903ba49b9af8d0de : SUCCESS in 1h 52m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/914ed9418ea24121b7ac17e96e6eb475 : SUCCESS in 1h 07m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e37d2ca2c2534ec9ba731e514b13426a : SUCCESS in 1h 08m 07s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e5ee6e8d47a459c9eccadc748ba3bdd : SUCCESS in 2h 28m 18s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea304a141481441e847f0db2f9fdf2e5 : FAILURE in 1h 43m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d82a576eff6a403d83d232cd5031f99a : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fdab0ec79df8c436fd9470a0fdedc8fdf0cf625e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-16 08:32:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"825af066624a283c7299b8daa47210203ffc9b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 11:15:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3330731d9ad9448b864856c96598d98f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e7a2141444a40379aabecd95f07f653 : SUCCESS in 16m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a7e533947a214547ae2672a3ffdd389e : SUCCESS in 1h 28m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/fa52a1088e8a4c8cb5993875548c9789 : SUCCESS in 1h 17m 06s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9067426765864c6eaa40a60065bc125c : SUCCESS in 1h 38m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7901bfb38d9e418387926279609308d6 : SUCCESS in 1h 29m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/354dd113c4bc4d26acaf9c3e55cd626f : SUCCESS in 1h 56m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/28dc847b6fd24a6891f5785ea880d1af : SUCCESS in 1h 37m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d43de23c304438fa42d3e6deb9e83e5 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd5791375b98451f8c8ef136dd0ec9b9 : FAILURE in 1h 26m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5bb39021b26e4f72891851320588bafe : SUCCESS in 1h 50m 05s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ccc3c05792f04b9fa06f2c39e6b36cab : SUCCESS in 2h 33m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea713d91488d46129583c4a80e5f08b0 : SUCCESS in 1h 52m 51s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceecfbe9944f41a19a4a7eeb246d95b9 : SUCCESS in 2h 02m 43s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5444fb34a6794a6590bd11d84573e573 : SUCCESS in 2h 17m 15s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e1ae847f61f4b74924e4d2336416fe2 : SUCCESS in 1h 48m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f816d50f4084f2c8e1689fe533aa5b0 : SUCCESS in 1h 16m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd45bcfc1acc4cc89d323353e63a26d0 : SUCCESS in 1h 10m 39s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f578a78d5a2469cabf26167e7a824ee : SUCCESS in 2h 33m 54s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e55cb7c1f1040089b67ad71696cdbea : SUCCESS in 2h 09m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3b39fb58bb34dd3bbf0fd5c20e01fbd : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"10ff259cc0210969f5b96afbfe5b6b8a9f055556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-16 15:33:36.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"90a825c579534ded7ea3295e1b6531d3a747b1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 18:09:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f9ab731ac2469eba286a375a4ced2b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41bc889ec6004dae8760ab9aa0cd2821 : SUCCESS in 13m 50s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/acec252be2d8415981ddb88e43bc8055 : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/11c874ea55bb456ab53c029ffda2eee2 : SUCCESS in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/12e88a0edf9c43efb55a805f8c0e590d : SUCCESS in 1h 30m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf17392192304e409ba7706d4ebece36 : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c5f99741c1f4a2086d1c0f49ba257f8 : SUCCESS in 1h 13m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/712f5e032e7044c6a2c23dcd22de99f7 : SUCCESS in 1h 40m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/320d0d33d12b4cab99399f6d11761a8d : SUCCESS in 1h 23m 15s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05dcb091f7d34f6db4a9eb49ba0bd1f0 : SUCCESS in 1h 22m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/02dd9cd25e2545f4a1b5ab1a06c1a9ae : SUCCESS in 1h 56m 55s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/90094bb94f944b75af8ad8fad594a6fa : SUCCESS in 2h 10m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3e5c18e796445a2a8c7a87b728679f9 : SUCCESS in 1h 45m 43s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54ec7aa6615242e98867184dbd3d56fb : SUCCESS in 2h 20m 23s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/89b6ef8c15c1465c9e381e3c1052844f : SUCCESS in 2h 22m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eab599d9c1274111a3c534850346eaf9 : SUCCESS in 2h 04m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3576f2a94b66411683003d2a29131653 : FAILURE in 28m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c60a9d003dd54926b746657a285f695f : FAILURE in 26m 47s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be738a07227a42e0bf40056c158b4055 : SUCCESS in 2h 32m 14s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a46e2a2ee1b24053bc8bd0af2312fe10 : SUCCESS in 2h 07m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c7bca0d3fc477eb844be03e06a5fe3 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":2},{"id":"7e0705e3e16af2bf56b4540102d7cf06c05bc78b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-16 20:41:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"61028b045c053cd3fa613f9af21705437daae628","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-17 15:53:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"05f845d969af0887f36f51d81d4f97f9d2435c78","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-20 08:00:42.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3735177a674566a645bbe7897b1f2463b44e3228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:02:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df370f86688595891d0b72f9627fabb6313bbb14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:34:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba28adac699745ed971acf011d456e9d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6510596fa600485aac0a1cad00b6b267 : SUCCESS in 13m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fd6899730e314abea834cba9b208224a : SUCCESS in 1h 20m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/40ec02c3e089453aaf8f6a2f77ae3ec5 : SUCCESS in 1h 18m 39s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a073e792cae44cbb8d1c75a0efd54b66 : SUCCESS in 1h 25m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06159b837a004429823cfda0e2d6f96a : SUCCESS in 1h 14m 00s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b6bb3c5aa5e45969c43bdf035ef2ccb : SUCCESS in 1h 13m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59ba5011db35480296fd525ab95127db : SUCCESS in 1h 25m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f291cdee49f2430da042e3f8c0f67405 : SUCCESS in 1h 34m 49s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51e776c9a8624626b28ef099ba71e823 : SUCCESS in 2h 30m 14s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58fee496674d4b67beb994b41c39938f : SUCCESS in 1h 57m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ded059071c4e03a7f31c0bb59161e5 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"9dc5922283ff502fd0ff35e720a05be83444b6ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:34:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2a01f9b9868831c94ca61f201f28692b74eabfca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:35:06.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/063b45c4499e47c29bbcda4444e24363\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdcb434dda61406fb6fea537773d741b : FAILURE in 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a649aea8c39b443b19ab6ed7580ba53336398cf","revisions":{"18b5e5fd73f9725f5eee9793bd18375baca29431":{"kind":"REWORK","_number":1,"created":"2024-05-15 08:16:05.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/74/919674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/74/919674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/74/919674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/74/919674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/74/919674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/74/919674/1"}}},"commit":{"parents":[{"commit":"58b9deb05cc59af5e2fb4c3ce40df3f483b81ba8","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/58b9deb05cc59af5e2fb4c3ce40df3f483b81ba8"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:09:09.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:09:09.000000000","tz":60},"subject":"Add tag to enable targeting of post-install config elements only","message":"Add tag to enable targeting of post-install config elements only\n\n\u003cservice\u003e-config tags are quite broad and have a long execution\ntime. Where you only need to modify a service\u0027s \u0027.conf\u0027 file and\nsimilar it is useful to have a quicker method to do so.\n\nChange-Id: I6f4d5f3388b71ce874650558b2566f795361a6c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/18b5e5fd73f9725f5eee9793bd18375baca29431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/18b5e5fd73f9725f5eee9793bd18375baca29431"}]},"branch":"refs/heads/master"},"6a649aea8c39b443b19ab6ed7580ba53336398cf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-05-16 15:33:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/919674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/74/919674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/74/919674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/74/919674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/74/919674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/74/919674/2"}}},"commit":{"parents":[{"commit":"58b9deb05cc59af5e2fb4c3ce40df3f483b81ba8","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/58b9deb05cc59af5e2fb4c3ce40df3f483b81ba8"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:09:09.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-16 15:33:36.000000000","tz":0},"subject":"Add tag to enable targeting of post-install config elements only","message":"Add tag to enable targeting of post-install config elements only\n\n\u003cservice\u003e-config tags are quite broad and have a long execution\ntime. Where you only need to modify a service\u0027s \u0027.conf\u0027 file and\nsimilar it is useful to have a quicker method to do so.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/919816\nChange-Id: I6f4d5f3388b71ce874650558b2566f795361a6c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6a649aea8c39b443b19ab6ed7580ba53336398cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6a649aea8c39b443b19ab6ed7580ba53336398cf"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
