)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~919872","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2F2023.2~Ic1e190fc86bd3b6fd2cc771273e28b5bb095a322","project":"openstack/openstack-ansible-os_cinder","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-23 17:58:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic1e190fc86bd3b6fd2cc771273e28b5bb095a322","subject":"Allow to skip volume type management by role","status":"MERGED","created":"2024-05-20 16:44:52.000000000","updated":"2024-05-23 17:59:37.000000000","submitted":"2024-05-23 17:58:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919872","cherry_pick_of_change":920014,"cherry_pick_of_patch_set":1,"meta_rev_id":"4409a94cb4df4aaf41c94201ccaa9d522891ec59","_number":919872,"virtual_id_number":919872,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-23 17:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-22 07:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-23 08:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-23 08:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-20 16:51:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-22 07:15:06.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-22 08:18:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-05-23 08:30:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"ca9936fc0da9c4b079be716c078351fec10bc3ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-20 16:44:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a119ef044e21f143019d5fbd71d2c80c1120b503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:51:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67be3b3994ee47b988cf5c6a338dd004\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c85837f6095435fba7d8e596809c527 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfc4145814fe4051b9fa6b0a5126d33f : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c75532e5b75cd55e81bf106a6e1881a30bff85b3","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-22 07:15:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab0652c4c24d48b189d978c8a8251cd83775d290","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-22 08:18:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"281e4899a431d241fd34809d5aea8ed2d54d38eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 10:54:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c543ac3be046429297d296819709e291\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ee7036859e04f919e875bf2e6282af1 : SUCCESS in 11m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a1362b83493b4200b8d8a865bced2906 : SUCCESS in 59m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/fbfe97696b864959b994e29a490f842a : SUCCESS in 1h 08m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bd6f0ea21ba4a08b588312316dfdc34 : SUCCESS in 1h 23m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/65659812013742d3b4ad4c9107a66433 : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dda6a47c2f5b4f088b378c3239bab84a : SUCCESS in 1h 15m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/997a2fd6307545b59cf8a26209134b0e : SUCCESS in 1h 12m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19243c0e18ff4588a632db3660293b43 : SUCCESS in 1h 10m 30s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63af012396684a89b83cf870c60e012b : SUCCESS in 1h 07m 58s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ed95aebbb9dc4022a3a04bc888f25f8a : SUCCESS in 2h 32m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3b4dda1c4ee45ed8d9128ad79b3b4ad : SUCCESS in 1h 36m 23s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/607f04828102404c935894ae6805ff4d : SUCCESS in 2h 19m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c88d5172703455b96873cdbe705ec4a : SUCCESS in 2h 08m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4d0afd01b5f64547a3cb277f9b3ecfbb : SUCCESS in 58m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00a1111b0bfa4bf6ae10efa0f8ba2759 : SUCCESS in 48m 07s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a855ba534049b39e545c39c6791626 : SUCCESS in 2h 12m 55s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39656c786a634448a89a71f076032fd0 : SUCCESS in 1h 52m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31faab14ae844a00a58ab56a0cb6e45d : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc75d51ee52643b38f03475da4f3079d : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c3a491ac471db67f3ff83e0820d7faee55a112f9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-23 08:30:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e04039941e6a5d037bb7dc488a9d83255899a6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 08:32:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aac6e177da367b67d7153d52e89d1efdd3ff6211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 10:46:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a52213a9d3f4c20bcd7b6e53fc371c0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a0f2b1bf61849e1889349ef22a064e2 : SUCCESS in 18m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a22d70c8d58640c99cf4d982304428e8 : SUCCESS in 1h 05m 07s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e674dd41ba454e4ca4499f72562f57e2 : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf703cd9810c454fadd469443f6cc278 : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2992aad816224a5e9f385827ab186c51 : SUCCESS in 1h 17m 38s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13d9f9b85ee7401eac386129583a04a7 : SUCCESS in 1h 09m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0030339614c0454099738ed5c82add50 : FAILURE in 1h 08m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78415346465a4dbb9cf89909ae2d402c : SUCCESS in 1h 16m 49s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ff5c9684636442fa6ca8be70647cbe0 : SUCCESS in 2h 08m 24s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a590f57aa4f747ccb70023da3dd47a84 : SUCCESS in 1h 31m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b104f825cdcc4ced925256a4e33e6545 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2cf3d10af5d470a94ea642150d8bdf2 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ab73e7829ebcba352281f21a47a724a50dffe8e8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-23 13:00:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5aa3e37fd2bed27093220c9b2e717a5c87193258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:47:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c25dff4c0c9443b4b7db191db1bafe44\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d79971738124b0fbf0cb16c6d06aa63 : SUCCESS in 12m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9254e41ff3f94c66872fb02fed11194e : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8d659d1fd371440d95fd7a2005f18aa6 : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c74d37f7f1bb4a70875e2a84d02bb413 : SUCCESS in 1h 15m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/adc3cc89f9ee4605b24d41dc69c7d5c4 : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34eb69ca4a2a4dbc989a0e5e1879b883 : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/01c5aeb5fbb2431181f501cf7f08c7ec : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea504183045549d984ba7fd4f5d490a8 : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c00c6cb9d25440929ba5042d7aaace8a : SUCCESS in 1h 10m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bbc6086cdac14fa6949389010cec9c13 : SUCCESS in 2h 23m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06a0cc9147e94b1d8f3975696766b48c : SUCCESS in 1h 37m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3de3f1cb9b7346698fd696ac085380dd : SUCCESS in 2h 37m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00afda4041034fd6a102da69534a411b : SUCCESS in 2h 04m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dbfb1992ba1342718cd06c14c8a7d86b : SUCCESS in 59m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3273a1cd6764e978796e4694b48bc3c : SUCCESS in 1h 07m 03s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b11bf870680743f09381991ccab502ec : SUCCESS in 2h 19m 25s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/991363c58ab74458aff04bd860fb1f4b : SUCCESS in 1h 50m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462230197f38405cbc574d9e001a1ba3 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd15f05182784277b45a4cec4c6cc2e1 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e927e1d6e3f420c8d58989c049b3d872ca7c5071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:49:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ecb0c1e7b254fae31c2ab6be695189011c9390c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 17:58:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0137c3d0a1784cd99cb8400ea84c4ec7\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9bdf9069f10478a9ab8fa856fd0207c : SUCCESS in 11m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dd79abc58ee642ea9234cf40285bd0e1 : SUCCESS in 1h 06m 35s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e7913895a1fb4354b9275e9c179df5c2 : SUCCESS in 1h 12m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a1a97e42e8eb47249e6379a4fda70121 : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f55e610854f4ff3beb42936f947f54f : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4227e5aeb5ad48d98237d1e2affc7973 : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a2ccd87cd16b4738908d85410e13fe2a : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5c587d755c9406b81a62d32c5121aec : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b95421d2f884faa9895cd76bbaf0f5e : SUCCESS in 1h 57m 44s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/323fa821ab5246e98486030ccbdcd0fb : SUCCESS in 2h 00m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca75cd5b76244d1da05b2124d598574b : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/875c4e5277e748db9ed3dca39db28605 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7cf85fc1c8ea6c806466ede177f89b124011fa4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 17:58:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4409a94cb4df4aaf41c94201ccaa9d522891ec59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 17:59:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e2168e32dc446b6a5f869ff29e3a57f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec4b508abb644f38a9ea122adef1ca1 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92f2434b17c8cfba8aeccfa994ccb4520c598157","revisions":{"92f2434b17c8cfba8aeccfa994ccb4520c598157":{"kind":"REWORK","_number":1,"created":"2024-05-20 16:44:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/919872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/72/919872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/72/919872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/72/919872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/72/919872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/72/919872/1"}}},"commit":{"parents":[{"commit":"7e33d2fbeb83f05ffeee249428d2d30999f4f9e2","subject":"Restart cinder-purge-deleted service only on abnormal exit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/7e33d2fbeb83f05ffeee249428d2d30999f4f9e2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-20 09:08:02.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-20 16:44:52.000000000","tz":0},"subject":"Allow to skip volume type management by role","message":"Allow to skip volume type management by role\n\nRight now it is not possible to avoid creation of a default volume type\nwhich will have the same name as a backend. In case operator want to\nrename volume type or have a different name - it will not be possible\nas of today.\n\nWhile we should migrate to using openstack_resources role for managing\nof QoS and volume_types [1], the patch/modules are not ready and will\nnot be backportable anyway.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/906372\n\nChange-Id: Ic1e190fc86bd3b6fd2cc771273e28b5bb095a322\n(cherry picked from commit ddddfde34d0cee46d2261eee638ab6589a1fd52c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/92f2434b17c8cfba8aeccfa994ccb4520c598157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/92f2434b17c8cfba8aeccfa994ccb4520c598157"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
