)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~941286","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I7220eec2957ea2ef8acfa9dba3e443ab39251646","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/fqcn","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-03-03 18:05:43.000000000","reason":"removed on reply"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-03-03 23:42:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7220eec2957ea2ef8acfa9dba3e443ab39251646","subject":"Auto-fix usage of modules via FQCN","status":"MERGED","created":"2025-02-11 18:31:23.000000000","updated":"2025-03-03 23:43:39.000000000","submitted":"2025-03-03 23:42:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941286-osa/fqcn","meta_rev_id":"c3e4c8dc5c6d23fcfbacf50653248016b8eb0859","_number":941286,"virtual_id_number":941286,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-03 23:42:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-03-03 18:05:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-02-25 12:43:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-27 09:19:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-02-27 09:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-02-27 09:18:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 20:57:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-25 12:43:03.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-02-27 09:18:31.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-02-27 09:19:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-03 18:05:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"9793801fc005f77f6ea1ebf9a64c873ca1e54da6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-11 18:31:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfc765aed4112cc1a1f73613d00db906116cae84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 20:57:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f732d764c5e4781b8bc5f4dc5c637a0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84de29072e6b436c9da569ee3ada7f6a : SUCCESS in 4m 53s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8d8549e78f4b4999b05c89e71311d599 : FAILURE in 26m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2850e006ec5f4d95acb45c6bcba2c97b : SUCCESS in 1h 10m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87dfeb12e1f04bcfbedce066452321c6 : SUCCESS in 1h 05m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83f3a830af814a34b01b8738a63908e6 : FAILURE in 17m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c03cab711be448b98af176e66caae0be : FAILURE in 18m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98120b01d4b84516b2e6ce9ed8fcc809 : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47a5a686302745ad8a7ca828c39c9a71 : FAILURE in 19m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd22a6aff9f4422188e5a60f9290feeb : FAILURE in 23m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/20641aa7acfc45f595156a8ba1837857 : SUCCESS in 1h 58m 59s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9008ff7762584aacbcbfcc29b9dc2683 : SUCCESS in 1h 56m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5189dd73311f4d87bdf626e45deedefd : FAILURE in 17m 16s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a62001b6cd6b4c28b483967eae03e520 : FAILURE in 22m 15s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b244afd2ef0e45f3be2edc866b95007b : SUCCESS in 2h 06m 56s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9854e9483f274850a17f365c485eda6a : FAILURE in 15m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/52ec4d3fa344412eb67354d8e045bde2 : SUCCESS in 1h 02m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e364bdca2377416e9405d940ce5720af : SUCCESS in 1h 02m 42s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/547366dd38f54213a4cc9cd23dae7d51 : FAILURE in 35m 11s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e227ab5c178143d39d0b42a89723e06d : FAILURE in 49m 36s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43b0a2e261f54d3e8252f80e23b65676 : FAILURE in 35m 12s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a3105f1a1264000a31b84e0d15f6882 : FAILURE in 32m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f8d3c880b648ccb1f6d98d6368a6d5 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"68985ab392e1fdfad15f1b1e5a0d8bd536d66fef","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-12 07:24:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84f0285b9c0696015616a5767ab04b1d982d1b67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 09:39:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73b83f4ec24d49bea33b62cb707a5041\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55c1512c077f4d5bb0e47145a3ec77e4 : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d8ecb857b68f4bb1b06d7d906f58f800 : SUCCESS in 1h 27m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/375dd6c5ca264072a0e9260ddc276c09 : SUCCESS in 1h 36m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ceef4b6b70c04b6ea230a189710ae699 : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fba4832c3e554a9baa54daf82cc8e86c : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd15e6be85794a709e7bf9681d07838f : SUCCESS in 1h 36m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d454a69436714461aa4cbc8f29e438f8 : SUCCESS in 1h 33m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e09c8d43e4114cd7ac22ac25ed84a449 : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7e7b411807241bdb6861d7b1b94ca22 : SUCCESS in 1h 22m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fcd635b5a5fa43c7a0fb13b62820df36 : SUCCESS in 2h 00m 09s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/153e63d0f45748d6a9455499b398095c : SUCCESS in 2h 03m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc7d4279004f4e5fb419667a03a685df : SUCCESS in 1h 50m 50s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73a5b88efe96430689488871ea7705d5 : SUCCESS in 2h 01m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/795ea87865eb41c5ac13da69997323ee : SUCCESS in 2h 01m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c03840736d3449b1894961a7fa496fb6 : SUCCESS in 1h 50m 23s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/04dda794190148a3bb5d46716d381479 : SUCCESS in 1h 00m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb8690aeafdf403e961288cd9953dab1 : SUCCESS in 1h 06m 48s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78f5f5eb72b34ab8befdbfd6c597767e : SUCCESS in 2h 05m 07s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/68f5727911e4424ead2509f03d84f1f8 : FAILURE in 48m 20s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17c33936ddee42d1bcef7dcc512f1eb3 : SUCCESS in 1h 07m 21s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8925bdef71ed42399e291bd15189048d : SUCCESS in 2h 03m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6125dda0cc9b461786665442ee55ce1a : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7d377dcae0b693218c3af750223da18aff3159a4","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-02-25 12:43:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82068c4407f2885529479bfb27e212f13176c9d0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-02-27 09:18:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbdc2abab27902bc715da6f5d6f9575e3e904b49","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-27 09:19:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f41e18ddf4e44377af5baf8567d4bb52b0e95a33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 09:20:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abeedb7c6f301e982c487b02345136d19320c04b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 12:47:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67e5e076012147c09763b3a7f8b73b20\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4deb51ed7994249b403adf019af1670 : SUCCESS in 8m 32s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f55a1d1c8fc7418ab80e529943653d75 : SUCCESS in 1h 12m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5c23a0ca0c3142668c77fcca83ee38d8 : SUCCESS in 1h 51m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37329f5e67fb47c0b1dd69dac4bff992 : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9270b02242324527837f426cb63bebd1 : SUCCESS in 2h 04m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0a2acba057744ec0a86518858045f928 : SUCCESS in 1h 24m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f54046b49554233bd617a89620720e1 : SUCCESS in 2h 08m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bc52221478c4339a4a933424ff2a843 : SUCCESS in 1h 54m 23s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbc3ba76383941d5a7aeca8c9b8795e7 : TIMED_OUT in 3h 05m 56s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fdbb8f801be4a079184af5568c3ca28 : SUCCESS in 2h 09m 59s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bfa330c256248f19df1912fbcf6fd78 : TIMED_OUT in 3h 03m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c63288aae3264b0bac445995f1a3194e : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a916414100c66931a7e873a17f0550c58f4ddca2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-27 13:00:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01b3bc1382ee3627cd0ea22dc954a033614dafea","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-27 13:55:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c3b44e836b00b08e0ce7d6a8f87a453fa516578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 17:32:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e2323142eff468f8ece3ddb1aed5b03\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7576722d66b4a1caf534080637c3b3d : SUCCESS in 8m 55s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0c3944fd49c54352836b24974d5d5eb0 : FAILURE in 1h 26m 52s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e64eb327c15f4fad9df3a524e08ef73d : SUCCESS in 1h 22m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4f9e8ff54594493b855c3c801665140 : SUCCESS in 1h 11m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50e0525a8396460aad59d4839ac0ba40 : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46e9cf38a13544e1b2881a2b17d00ef6 : SUCCESS in 2h 05m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/93986ae0194143a7a77ff4cf49b75e17 : SUCCESS in 1h 24m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de78b66b40be4d699228e528c4f76963 : SUCCESS in 1h 18m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e753aaf273474149948caff8c599dd9f : SUCCESS in 2h 05m 15s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d0282a0039848abaa91aeac29a3f286 : SUCCESS in 1h 53m 47s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/30cf0fb06cb44c3aad6e69c84951cdc4 : SUCCESS in 1h 52m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9222bf8c689341f7be7deda49e4d8de2 : SUCCESS in 2h 42m 42s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84b6b934ccbc4600908dd24c77b73157 : SUCCESS in 2h 07m 16s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eed5d3a3b63848ac92ceef1fb303dffc : SUCCESS in 2h 00m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8766a8c523fa4d7bba0e0cf452edf3ff : SUCCESS in 2h 07m 47s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/869f4d8088b24575a6962e6a5e2f380e : SUCCESS in 1h 02m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69a2b21809164c0abb37dff819eafbc9 : SUCCESS in 1h 12m 11s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03fbaf6421e44c03b46218d186d39483 : SUCCESS in 2h 10m 18s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d19a6c6f637d471e8aa6d8eec3cfe5ff : FAILURE in 58m 38s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cef17682a834061b78f071d41445db5 : SUCCESS in 2h 12m 16s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e6a54c581d84610a158723088c17039 : SUCCESS in 1h 51m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd39c34158df414b98a28b625a1edbf4 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a3fc6fa008f52c1c1488768df14aa9d9620d5ab7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-27 17:33:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d8acda2a87f5f1769d2069810bf275c0fd8cf6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 00:46:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/faf1a640a085494cb2b9684bbf53d3f1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf994cf91aee4f3587661be71a5f0c59 : SUCCESS in 7m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4edcb4e81bdb480b8a3104d3e6b8bed0 : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c6b8031f13544ddb1ec937f46d2c319 : SUCCESS in 1h 22m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2298740232a4113ae0b5f4a8c039960 : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/073c0e7a04414fdca4d921898aa0dd14 : SUCCESS in 1h 12m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2525d4ea77e546a5b3ba68d344010358 : SUCCESS in 1h 17m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/62fc5b4cc3514a0b832d61e0be0e09ca : SUCCESS in 1h 24m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb2c68f96c4c4792a9e1d7e071ed5e41 : SUCCESS in 1h 19m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9d2057f70f84bdb89d002ee4269e7a2 : SUCCESS in 1h 43m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/948a3afaaebf497e98ea5cf9a46d0254 : SUCCESS in 2h 15m 13s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/864191dd106947da8cdcf477a53456e9 : SUCCESS in 1h 56m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/549cbb6f390846c68f0d27253de2480b : SUCCESS in 1h 55m 41s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12613de602f143b2bb48b8c76b0d4e1c : SUCCESS in 1h 33m 34s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/332a0e788c0945a981e97745e4ba0015 : SUCCESS in 2h 00m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89c8cc2e5b124bdf87c157527dc07be7 : SUCCESS in 1h 50m 16s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39ff233169d145ed97cbabe42a780d58 : SUCCESS in 1h 05m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bfd4d7e8c344a088f4f074f0f7b1d4e : SUCCESS in 1h 12m 36s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4959d5a438b455d80648416543fb405 : SUCCESS in 2h 11m 11s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3483f98be06743bd86fbea36b8c6dcb1 : FAILURE in 46m 58s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4837dc4622174fdbb919234d9b9b0335 : SUCCESS in 1h 45m 58s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd0afce783fb4b928cace70a2844c588 : TIMED_OUT in 3h 04m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd332f641bd44ecabf26b67889daa754 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c16d78f9654963b20be909a73996d122fd11cfaa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-28 08:26:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d244e20fdb73966ad0cb44689903bbcc6207c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 10:59:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57acbed6b2954714a072073827429d80\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a9fbe06f38444c6873c70bd6710bc01 : SUCCESS in 8m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/561f15874fe6426b87b3132ac6cc6dc6 : SUCCESS in 1h 33m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd76161a813a4f71a8c2efcc501e3467 : SUCCESS in 1h 28m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73d81009cc77485e8cacc360ba8bf2cd : SUCCESS in 1h 16m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75eef16cfd894c9383cdf195ba6fe3c0 : SUCCESS in 1h 14m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8745e6d7d72f4dfea69dbda797d91f1d : SUCCESS in 1h 17m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a106509f0b0d4257b486d770cad2b34f : SUCCESS in 48m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/885b2459d2dc4c03b91d31c4d7514de4 : SUCCESS in 1h 16m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15ca15f318fe4f2cb7544f4ece05a4cf : SUCCESS in 1h 18m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2348686621994a5f9b02e2080b26a8c0 : SUCCESS in 2h 01m 42s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/669149293ab44096879a34d33954e871 : SUCCESS in 2h 00m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ed12d9104244a3da041735d58f49a83 : SUCCESS in 1h 53m 27s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ff64feb09fc443eaee40815feec6e3f : SUCCESS in 2h 09m 32s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/46d2efb257b741b38ca32c150fc6988e : SUCCESS in 1h 58m 15s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d164a97096d949dbb2c7c5875bf4d84b : SUCCESS in 2h 13m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3db42c58ec4449e9508666a136f84be : SUCCESS in 1h 13m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa369b3aedbb4a9da2a5d6cfd6a71bce : SUCCESS in 1h 14m 17s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19ad68cfb0484d8c9c8713db1997a857 : SUCCESS in 2h 22m 55s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/444c0e9fdec04b04adaf850362be18c6 : FAILURE in 52m 38s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce2779f8c3d34e909bbd8a6541558747 : SUCCESS in 1h 55m 17s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45f7f09ba28341b19e83bb2ea8c70b07 : SUCCESS in 2h 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7988156a6144f4682e90480cd73dfc2 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"816acf15983a9a27c2fe774977c3b6537374911a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:01:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d58c81b33ee4e47e7692b9444a17b3247d47f3d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 13:58:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99367b953d0c4c74b06cac515b9570f5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5ab5650246749ac84ac59d9e805292c : SUCCESS in 5m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/78a3df16d27647cc80f42141b117a16b : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/18cf2f0bb3bb47068ac8588feb184238 : SUCCESS in 50m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2c29b80adc3404a8a8e3c7f8457b0e5 : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43be994823a94e64bc3f5dc7871611b6 : FAILURE in 1h 29m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d5f5b3c3886442dc97f70dd95ad299d4 : FAILURE in 1h 14m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32897d63c85b4a04a7967f830d9ca057 : SUCCESS in 1h 13m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d70582f76fa44ca38783bd878f43e689 : SUCCESS in 1h 19m 39s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67e9e0bd3fe2458aac1677098a92bfdf : SUCCESS in 2h 10m 41s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c9f85053337469cb4c5ba18a1e55e6b : SUCCESS in 2h 07m 44s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbcfce3f6e714b39bbdeb4f40d71f1f6 : FAILURE in 2h 56m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b657539928c9493da8322eb3d622f705 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"797bf40f11e0fbe52bf0d5c15475ddaf8cf5494f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-28 15:07:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0c02671c3e1ee6db923cd869f4435f9fe4cfdd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 18:34:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2272828fc184f4ea6a5b6660847a4ab\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/703c382aac034af791ed3eb4b88c0999 : SUCCESS in 9m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b1ae3fd5939848c7be963c0bb0fb37ef : SUCCESS in 1h 09m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/71f34af635f94a8ba78eff1c2d71106d : SUCCESS in 1h 24m 03s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1dc1ee7f04e344b895bfefeac723fa27 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c34616673a74751afc112aa28e96598 : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f2766d7744d4ff88e68151eec6beecf : SUCCESS in 1h 28m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a41fdf4f6ca4c3c9356d8d860c4bcb6 : SUCCESS in 1h 28m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33d2bc33b9044a2b985861cd85e44b3c : SUCCESS in 1h 17m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8fab1470c104a3097952bd88f78b361 : SUCCESS in 1h 21m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87c0fe27116c401399f9f47d1947ef55 : SUCCESS in 1h 55m 01s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83e9ef80bc174c8b9ad37d2484420133 : SUCCESS in 2h 05m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/768e5bb8fcd7407299c4bc9c30171c6c : SUCCESS in 1h 52m 28s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/157db09fe2674fdaacfe6ac5d2abdecb : SUCCESS in 1h 54m 36s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7fbdeff76f8446959e0d6b9b0f18ac3e : SUCCESS in 2h 25m 57s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/533843d193914e9dba00a9b6cde9c70e : SUCCESS in 1h 48m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e1315c943f494f1b8785bec99e19ffa0 : SUCCESS in 1h 14m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cf845cf3afa457685fb4639fb42202f : SUCCESS in 1h 15m 50s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b53f1708eb3c495f921a0b97d35c1a9f : SUCCESS in 2h 24m 27s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/04e1b46e08f64935af271d9dcb927cca : FAILURE in 32m 14s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71515e71a76b4f6597c83be666dc5089 : SUCCESS in 2h 09m 23s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/718135fa7a3e4d6f82fca99d3a0f3b5f : SUCCESS in 3h 02m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4ffe0e440e4c96b4a6ab9897f22f3d : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ed01b54610a60c0558c4ea4e491d64127d1a66f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 18:36:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93b503bbe789b54e4c8a8f487ce7c48be1637375","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 21:09:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0e9a2ce582e4d419bb2d02230018902\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18d487b923114fc78a9fedaddd3935f3 : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1850688775094df3b05312ad1158b67e : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b51ee115158e4110ac2b8ea12f2bca5c : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25e14e23abaa4b0e834878e1e0b35560 : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66c9019c600742bf9310bb3c8f171cc5 : SUCCESS in 1h 36m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5470459244d943ecb1527f9a63d4a021 : SUCCESS in 1h 33m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2667de56c0904558857c1964cf352933 : FAILURE in 1h 28m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ff3f57ecb874f57bbf0929ea63b55a9 : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f688135a88ca4998b5f9f242ef3dbc22 : SUCCESS in 2h 30m 29s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6d943e5cb3846779e9e8ad51e4276fa : SUCCESS in 1h 17m 16s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea04e8e2f6f5473591b47777b031cc39 : SUCCESS in 2h 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd3b7aacd2d4ba98d5d5d23da9d6f21 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4a04aa4b57e1e74ca34d804b4c088e299654cbcf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-01 14:45:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0adb44f53ff7b93f3894a4c4b75d4d80b9612e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 17:08:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdd1a85e68b941e5869576492a45ef01\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67a828f15bc7491296e14ed84250937f : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/84b3e2df91d44469a509d780de41f737 : SUCCESS in 46m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/264c75879625496eae258d6e177e2843 : SUCCESS in 1h 14m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/69cc764735684f5cb7727ba3f2bbffe2 : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1187b9eebc24801af2fc1ef7f3f3014 : FAILURE in 1h 10m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88f11fbaaabf424d92ec4011dfde40fa : SUCCESS in 1h 29m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e4a5c1226f1547fb8ce9d86b17df2719 : SUCCESS in 45m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4e6c81269294aa3b7b36844e614b880 : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecdd369db7c743baa96cc62cfd3893cd : SUCCESS in 1h 09m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db2b3cd3163c4d13bc2d7858a6ea8dc4 : SUCCESS in 1h 59m 23s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c4439c12068f42aba7734d895cb2492c : SUCCESS in 2h 10m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/070940ebe8914aa3a7a53ca41faa546d : SUCCESS in 2h 02m 58s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd9e7bda9567456cb78bcdacbd272694 : SUCCESS in 2h 01m 06s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af5100189d574de3a74ffc8c0eac7d1f : SUCCESS in 1h 49m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28f07152c32e4302ab57c2e863b2db52 : SUCCESS in 1h 54m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0546643d3bd4c83a9268617c298f14c : SUCCESS in 37m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66e6ada89b884c77be7c0293a82c584b : SUCCESS in 48m 42s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fa2b7d50a69482293b30cd8d237bcd5 : SUCCESS in 2h 14m 21s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b72920d85a8f4869b1008b4f4e4f307e : FAILURE in 50m 34s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/872d139c10f24776a7814eb3517a1da8 : SUCCESS in 1h 46m 29s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbcfa904ee394d3fa6bf6f9f408429e5 : SUCCESS in 2h 02m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7319c0501a504180a0bfc53aec9f986c : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"76f6d0eee839c957bc1f740f64efe0abde65ffb2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-03 10:57:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb937c4ead92d755f409825d405a29d30f76bdc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 13:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db0aacfec114404d95b216fd5c51c628\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fa67ac918ba48e5b01d885adf041d3d : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b8313877c0824ca280415cb8470648a0 : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dae9b2738e84636965435df07007999 : SUCCESS in 1h 24m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9de161b45804c78a7f2fb3a55cfbb56 : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62a9fc8b4dba4a9fbd8b1fb443efcc10 : SUCCESS in 1h 11m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc16e5a5a1524f129a15de49deaf2412 : SUCCESS in 1h 29m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/684ce5bf810144fb9bab11af4bc4f9c6 : SUCCESS in 1h 14m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8305a60e9d7c4e988d8e3f2f23b691cf : SUCCESS in 1h 23m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfbab02e4aa9481fbb16d26dc78cdc7d : SUCCESS in 1h 29m 04s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a6867de0669e4b42baf81ef4b0a8352f : SUCCESS in 1h 50m 47s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/62350cf789bd4ecfb21e7391a0f40e92 : SUCCESS in 2h 00m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/740e5864f1fc438bb2225000c8d3a647 : SUCCESS in 1h 45m 56s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7d467963a8f44989ec3c915e8ccfb20 : SUCCESS in 1h 51m 29s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3c679f5f8ae54cdab0e3a32b4bf08485 : SUCCESS in 2h 11m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe32196638fb434d9324a52cb934f9c1 : SUCCESS in 1h 52m 03s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ddd387981b5244c2bcbdcdce87ac9ef1 : SUCCESS in 1h 19m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/468c36a16ee24df69ebec1d585edd01e : SUCCESS in 1h 04m 38s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef130e8989394bd1a9366e2f06fffb28 : SUCCESS in 2h 19m 36s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ecc006bdd2ee4560bb3b624fe784efbb : FAILURE in 45m 40s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce8bd653a853444790e4ad1fdd3a1451 : SUCCESS in 1h 52m 09s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fad06022e3c2494b967405368fe43468 : SUCCESS in 1h 56m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7e36afc6b748c78329048164abfcca : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0acb7af9c1daebdd1dbcc26ed2cdcc196c1014bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 13:21:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1631c325c4604c9f292b5cf23e26f0b426214527","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 15:33:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3847bb3f27bf4c54a62943cecf1fc156\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d6b0e551a634884bf2a397b34228b5e : SUCCESS in 6m 59s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1f09cc2863694d8b8d324ea57404ef1a : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ccc35c1e361f4bccaf316659bdb5511f : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4249dfaab70f490280d62ffc94416b4e : SUCCESS in 1h 21m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43603a1a1a2a4d54ab57eae2289c70b9 : SUCCESS in 2h 07m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/95aa5939d32d429f984dbe4a33b773aa : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b9a87d6291f4335808c6db1f0932773 : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25065956b48e4d86b80a8c56fe1ecc00 : SUCCESS in 1h 35m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8607a9b5fe694d9d94ff6615cde45e37 : FAILURE in 2h 07m 37s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dd7630431604c6bbfec7fbe47bd246d : SUCCESS in 2h 03m 28s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85fc64ddf0174a058d2494d98f0c3999 : SUCCESS in 1h 51m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3857954bee214b2f8eac06c8d6487b27 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"76b8bf1398bcc7e53414d1694871db49f51dfa4e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-03 15:34:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f437e0aff21992e2e639f9b40b8da1282ae862d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 18:01:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bd4f8ea63af44c9901eb2ce709bebfc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbae01cc0e404657b0368b2185c656e2 : SUCCESS in 6m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cc5fa6d372924c81a0b1b680c7601e3d : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b23a0d18c31f465cabb57d9996be4231 : SUCCESS in 1h 21m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6baa91923214176adc5b61139a3d182 : SUCCESS in 1h 26m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10f7034676024fa09dfad9dbc45e1414 : SUCCESS in 1h 11m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a27ee74246449e1a17962da133f1985 : FAILURE in 1h 24m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0fa555fe1cee4eeb887defdfc15e8179 : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4428c615ba604353a89be7604729bd63 : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/863eaa700eeb4f9ea779d5a7736c5a8c : SUCCESS in 1h 28m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c66fef1f692740c8bb34ae4510194729 : SUCCESS in 2h 01m 47s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/171e1f9a20994c0c9fa6b4c177b856bc : SUCCESS in 1h 56m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72d22744b02d47af87839cb09f18fc43 : SUCCESS in 1h 59m 19s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d4862e15bc34b37ba79f498d8079ad7 : SUCCESS in 2h 04m 54s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b412f0412fc7400f98c410dc5d398e81 : SUCCESS in 1h 49m 15s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d838bde5fdae48efb94a1b412e831d0b : SUCCESS in 2h 03m 54s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/88487b2537404c50b353501b3d39bb27 : SUCCESS in 1h 00m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db49a76985744e49a4743de84cc908c4 : SUCCESS in 1h 00m 45s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d4f46bce7944d3abefce34fa3f462c9 : SUCCESS in 2h 19m 42s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/760daef1749a4939a3f69281eeb5b8f1 : FAILURE in 45m 33s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/347a0d1f105845d4b69c922d5b1de3ec : SUCCESS in 1h 52m 43s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6689b99afa424b6a8f4c7ced740ad9a4 : SUCCESS in 2h 08m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1ba4c50a6d54638b4b99bc540bf10ba : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0743020940d0633c56fb1d5efedceee2f87c378b","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-03 18:05:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"14b72118dca073dfa6ccaec820b87d1057df7cea","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-03 18:14:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8784518375187fe6e413eb2134ac3b477f5d2876","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-03-03 18:52:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fc7a0a5a8566fe2fd4dd8c595dd4a240804e2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 20:47:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/072eb98c9bac4d93a97943a0d6dbf38a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/423af4969b9e40e192f24d7b955a6ce3 : SUCCESS in 6m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2db43ec4eea14b50ada6137707e953e5 : SUCCESS in 1h 09m 06s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8737e3b98af94b0e9e38462b8b7ff29b : SUCCESS in 1h 25m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3f2e30a4ebc24621893b9ae1dfc67e40 : SUCCESS in 1h 25m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55fca19a8c2f41cfb0cf2b07735a9514 : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79c1fc122919415f9a65643bbec6617b : SUCCESS in 1h 12m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0ad365a7bbe54c73b25b1d5d2ee97a5b : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2257472185c342579dafa669bc2971ff : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61a6039e3e064c3492e3f5b95eeac95e : SUCCESS in 47m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31d780ba297f41e6b9d4c91bb7868b78 : SUCCESS in 2h 10m 07s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ea1f86df41d4f45ba980c7132ec6540 : SUCCESS in 1h 44m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68890f7b485b4b5ca0d65b1a4b67376c : SUCCESS in 2h 02m 24s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48210c720a274de7870b6f9f393edeb8 : SUCCESS in 1h 58m 42s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e4b396094a4e4430a15bf191bc0388b5 : SUCCESS in 2h 04m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18cbba7227994000811b8422c13e0fe6 : SUCCESS in 2h 03m 21s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/49f8a12daec9470ab5838638f7ee3111 : SUCCESS in 56m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a8ac8247b8547be823493604958be6c : SUCCESS in 1h 06m 52s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e71ad26c5fd4ddaaf0814bb32dc9f8a : SUCCESS in 2h 23m 29s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64ea4358b3984ceeb5135c739ae55fd6 : FAILURE in 42m 14s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf79d8dce07b47cb9ddea205e737aede : SUCCESS in 1h 51m 33s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfdcac7bdda641b69d888f1e408e9a7a : SUCCESS in 1h 53m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4963fcd99cc4469aab613d662cbef0d : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"34ce51694afebe7cb953f97cd049599c9c552be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 20:49:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf82e24465316ede9075f12735af3c9f71936512","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 23:42:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bec13ff261449ff8efae023fb8dc088\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91e3fffb8b75433dba1b421e5535444a : SUCCESS in 6m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/53c49b4e63b74bf782362bb12c13c1fc : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/270730530247435da36e2750ba344ce1 : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7aa89706c2042108986ec9b375c6335 : SUCCESS in 1h 18m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/741c536f09534bdd91489857ea11cf31 : SUCCESS in 1h 15m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa12ed293dc647749045c1eeb3f48270 : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd141ff739bf49629abe51393468f05d : SUCCESS in 1h 18m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd9a19f71f664a729409aba199f9b2c5 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd8ad3c78bd5427c9e3ce419050e4fba : SUCCESS in 2h 48m 29s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92f873e7b8dd4d31b4b1b535a00fe6e8 : SUCCESS in 1h 53m 52s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da68469cee02435996a06652565dfac9 : SUCCESS in 2h 09m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff9d320c2ee4da4ab71adab4be6cb25 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"778f5ed5aa11368f9df147a28f186dfe500a3325","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 23:42:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c3e4c8dc5c6d23fcfbacf50653248016b8eb0859","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 23:43:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21288836bcba4cc5b1398762580b10f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0408dddd306648e393cadc543f1caba9 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4416d82c8927e61971e9bb50cfc171076633e2ab","revisions":{"4416d82c8927e61971e9bb50cfc171076633e2ab":{"kind":"REWORK","_number":1,"created":"2025-02-11 18:31:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/941286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/86/941286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/86/941286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/86/941286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/86/941286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/86/941286/1"}}},"commit":{"parents":[{"commit":"014a97fe88b35fb227a9e5a01e55420022edf3c1","subject":"Auto-fix yaml[truthy] rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/014a97fe88b35fb227a9e5a01e55420022edf3c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-11 18:31:12.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-11 18:31:12.000000000","tz":60},"subject":"Auto-fix usage of modules via FQCN","message":"Auto-fix usage of modules via FQCN\n\nSince ansible-core 2.10 it is recommended to use modules via FQCN\nIn order to align with recommendation, we perform migration\nby applying suggestions made by `ansible-lint --fix\u003dfqcn`\n\nChange-Id: I7220eec2957ea2ef8acfa9dba3e443ab39251646\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4416d82c8927e61971e9bb50cfc171076633e2ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4416d82c8927e61971e9bb50cfc171076633e2ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
