)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~942581","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I7a8e563fa60a62bdc24d5a3c9d7fa1a43f324b76","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/remove-cinder-v2-crud","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2025-03-31 13:04:33.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2025-03-31 08:48:28.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-03-31 11:07:38.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I7a8e563fa60a62bdc24d5a3c9d7fa1a43f324b76","subject":"Align on cinder service naming","status":"MERGED","created":"2025-02-24 12:56:55.000000000","updated":"2025-03-31 13:05:55.000000000","submitted":"2025-03-31 13:04:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942581-osa/remove-cinder-v2-crud","meta_rev_id":"9a7f94ab7ce48de8a1194e1fd28d5078d986829e","_number":942581,"virtual_id_number":942581,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-31 13:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2025-03-31 08:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-26 11:17:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-03-26 10:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-03-31 11:06:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-24 15:02:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-18 10:46:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2025-03-26 10:44:23.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-03-26 11:17:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-26 11:17:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-03-27 15:20:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"53547f042c0ff65319ea12639df15a1a2f423b92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-24 12:56:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f548cd2502d7d4e804768260d6c82ead0cc641fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 15:02:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cce28af9205b498781741efaeb26b514\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b22b9889bed843d0bd0d2833f53cfd47 : FAILURE in 6m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/45c6148055f84aaf98b7f5d1b13ab315 : FAILURE in 44m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/353404a34a944298b722a5ff3eb323e5 : FAILURE in 1h 20m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5303fa5e859c4519905971c05ec2f5d4 : FAILURE in 1h 27m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/129d416d474743da8b91cdc8c97e3b34 : FAILURE in 1h 12m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29510b67567e4f578339b55dfbb9fdd3 : FAILURE in 54m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39d649ba8aed49d3b2061770187f0e75 : FAILURE in 1h 16m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c06400186014477b013b213eae932d2 : FAILURE in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/024ec61610d54e0fa8a5fec04b56e095 : FAILURE in 51m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb0bbd78d4f74ceeb8ff1a92d1e7a35a : FAILURE in 1h 04m 53s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e29b8cc9f664813b7b0930a0cc6b18a : FAILURE in 1h 09m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f5316c1138464d9eeb48bed8e9d74b : FAILURE in 1h 12m 46s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/561471756800464b9b409736533429cb : FAILURE in 1h 02m 24s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae959ed6e0384f8bb23f51747780682e : FAILURE in 1h 07m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/192847e9a59e4d5d9bab48750e6674dc : FAILURE in 1h 02m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aed1d6d15e7a4cf894242914ee62cb73 : FAILURE in 1h 00m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e55388b57a14e359d933497254aa456 : FAILURE in 1h 08m 08s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/280595b67e0f4e21b6406c8e747ce407 : FAILURE in 1h 44m 12s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d5b400e44b9448f1b1cdf1f4da7988f8 : FAILURE in 50m 28s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f288d56ab1741c69453c2cb56d1c7f6 : FAILURE in 1h 36m 23s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/393409400f2f4cb68c9853efebedbb4f : FAILURE in 2h 02m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dec33066b5b449e86bb11c847850c48 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdb04289f2f9453992f48dc18fdf5e5f : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"352e91054c9056b5027fe5ceb36679d8d5219142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-26 09:50:39.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"717ffaa1dfce2dd2da419b70853151c35e2182b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 12:23:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6be4c25a5ec241f8b82d2a8f055b42c3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db74d27dfdf34eb78497dee6de44295e : FAILURE in 6m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/29b8f19b7def49fc96f4fbbc7ac7ae6e : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/559f45b5b7dc4812bf0aa391a41d98f9 : SUCCESS in 1h 21m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/08bfd44a817d40cebfafff2f5b4ea636 : SUCCESS in 44m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83ddc14d10164cb1b116b4d14cda9bf5 : SUCCESS in 1h 26m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e450459767e047a9a58e920bb83adb5f : SUCCESS in 47m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b6f804050eb4320b395d74cd70b0bf9 : SUCCESS in 1h 19m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb05172881fc40ae8a2f943e06565875 : SUCCESS in 1h 13m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7590ebc4319a4d4796cf2a5f65c2bf7e : SUCCESS in 1h 27m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a7f17cdffaa478791d771d2d38e5540 : SUCCESS in 1h 55m 24s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4db6a530835341a5bc76e401c4b22249 : SUCCESS in 1h 53m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aab334ee5ce840549f4bf7bdcaf58d49 : SUCCESS in 1h 56m 51s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47a23b92739b4a729a5c2fc5bbb45022 : FAILURE in 56m 59s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/09c297e56d714b0bbdec8776025eb1f4 : SUCCESS in 1h 58m 50s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc7b43d489f041fdbf88be350e04604a : FAILURE in 1h 47m 10s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4ec394d297de4376ab23fbcb2bc8007a : SUCCESS in 1h 10m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd7299d712bd48f3b8a9940f96862e23 : SUCCESS in 1h 05m 38s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03bf59ce13644e4593bcae5319ecb54b : SUCCESS in 2h 25m 48s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b63f0ce7f5474e08b64b85d2198d0869 : FAILURE in 45m 12s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/006542e6c5c84b0d8e222111a653b4f1 : SUCCESS in 2h 10m 14s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/783c54a4dda943438aec22bffdc02137 : SUCCESS in 2h 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db02067c1ad44ee93f79e0b958fd5bb : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2359fbc0bbe84d5ba245e97100b97e7e : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"2c32ce74906e31b08386bc9b9bdc6defd80544d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-26 12:47:02.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9f1eb5877eddeb4b1a7a8b8990b4f3ebbba63dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 15:00:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdfa01c8d1dd4498841384162f8510f4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5f5dc033ab34e619e150c7a2d54c0ab : SUCCESS in 8m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/db09583213644fe189ae82a3508be7ff : SUCCESS in 1h 38m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f63769e3b24246d9837f247348fc0677 : SUCCESS in 1h 37m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfd3334092104ccba577243cf1d1239b : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bccaa8272ce44d558009fb451ec94839 : FAILURE in 1h 12m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/709c5b61870c430087a321bcf6b6fb9f : SUCCESS in 1h 13m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d7a095233f04ffd982d4456d17c1318 : SUCCESS in 1h 22m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbe75f1638d743fea393f81357815b02 : SUCCESS in 1h 16m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aec5c540d49040eea022b41162eed3e7 : SUCCESS in 1h 31m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6da3e40682cb4ef2866448f4c709aaca : SUCCESS in 1h 49m 53s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fe6e57ec7e0b4006a8460f10117739ec : SUCCESS in 2h 05m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4d5e9921c8349d983a9f92bee3942b5 : SUCCESS in 2h 02m 28s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7ef4de919e5480bb9e3d8f32ddbe706 : SUCCESS in 1h 09m 01s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0c57f6e50d984cbfbc16f71f1b0dc6f0 : SUCCESS in 2h 02m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b930ec80aacb4327b0a44ca48bd24d6b : SUCCESS in 2h 05m 19s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/112fca0eea64461d8ea4d16ca5e18fbe : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14d975091da14e1799f0ab56bfa8153f : SUCCESS in 37m 13s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99551ebbe6534e3cb6bd8b891d45a3fe : SUCCESS in 1h 57m 27s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ed7221a140bb4edc82edb723ce4ec530 : FAILURE in 54m 02s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bee5b4c50d6243ffb881cfe7453fc030 : SUCCESS in 2h 00m 29s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9b131345e4c4f50bd43256bb586c6fb : SUCCESS in 2h 05m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40dcde6c61e34dfa873b6b408ccf23b2 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d4f10b7ce07442a90246b0889454a74 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ddf4852cb743db4523ccc2d866b995c1e569953e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-03 10:57:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4e52f29b244bf985232d1b962fb57bb55b625106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 13:25:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44bd1da9536149a5a55ef8cdd13a6878\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef96161b2e8e49c5969b8713efa23f46 : SUCCESS in 6m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3ed43212928e46d0b9f4a298ab2885d4 : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/203c5ab4852a4220b482911a09bffdbc : SUCCESS in 1h 28m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/40b0217169e94e58a872a3589e311449 : SUCCESS in 1h 31m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afcef6f81a45435fbd32d43226ece8c0 : SUCCESS in 1h 23m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/497c5be9287f4c058d81aed53869f6f9 : SUCCESS in 1h 28m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7bfa77cf49ec4f3eb370615482798fd4 : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2a9e9d5f209474b9dc068efc41dbe5a : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f7ec7d2febc46dd8cb58aec130b1faf : SUCCESS in 1h 35m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db71975decbe4ce1b1d1ab42dae01a7b : SUCCESS in 1h 46m 25s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b7c87456458d4149b25817c8ecc4f81e : FAILURE in 54m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b26f6bcdf8940139085843dc121b2b3 : SUCCESS in 2h 06m 43s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/569f9a8cc080448fba6d4f19b0fd1db7 : SUCCESS in 1h 51m 18s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ada7203ce28448179c29727b906c6f4f : SUCCESS in 1h 43m 56s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d68f94843aad4c9291af153aa30027de : SUCCESS in 2h 02m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a39f046866474352bfd595668ac368af : SUCCESS in 1h 00m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/909587b0d4be42d9972b268c2beb47c4 : SUCCESS in 1h 17m 42s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1521bed1ec6f4f139569cdb0411081cd : SUCCESS in 2h 12m 15s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b05e05aa662241548af20c9f343a8130 : FAILURE in 49m 57s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7a8bffd75f948b4858ab75e5d894eef : SUCCESS in 1h 37m 01s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7080b2ba077842a7b06c9e74bea4065a : SUCCESS in 2h 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13cfb2705c0a45d0918b7243a0d8e046 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4951f94873204f35853b18acf56db118 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"b0663143c06b7d24090a4365a0d9a2fec58c0dd3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-03-18 10:46:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82ea753403570463038ba304876a18d67b85c31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 13:13:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64b2bd2d7a9b4aa98c414f0d91ce2f49\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c359ab20edb4f5f836814f936172d39 : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4a6842ae36f944e989bc26737497187e : SUCCESS in 1h 04m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/32feb448a25c4a5e9bced702297bdd35 : SUCCESS in 1h 17m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b949123257044d95af36acd2e5bd16e8 : SUCCESS in 1h 07m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed9120042caa4936a27151d9ae57c9f7 : SUCCESS in 1h 23m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bea4cc2a3e54dfcb43a028df76ab08e : SUCCESS in 1h 14m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a9850299a004475b9b4c5a8fda4da1e : SUCCESS in 1h 27m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2976192109e840d793cbbb7932cc8b1f : SUCCESS in 1h 11m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ff8663a968e48f3a1076868807fb28b : SUCCESS in 1h 12m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e9333e74e9543f5bc641f44cdf44acb : SUCCESS in 2h 02m 36s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/97c5e84ba6594a329e4352b98ef92868 : SUCCESS in 1h 38m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30be9752c87e4929b506ec7c3300d418 : SUCCESS in 1h 53m 16s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53d8e3c60a584b798f9152f019e3bd8e : SUCCESS in 1h 33m 50s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af17839bf86d4259a200c7b72d9f23e0 : SUCCESS in 2h 03m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/361742aa14874286ba0bcc6703afddc6 : FAILURE in 1h 37m 28s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/15161b994b114f30a66bc2997e8844de : SUCCESS in 33m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e032dc7867e649d682c33298eb037c77 : SUCCESS in 1h 09m 18s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be06f716ce5e451687abf84059264b8e : SUCCESS in 2h 17m 22s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e7b6b349c89496f9338d2764a3b9322 : FAILURE in 41m 35s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b16747047b048469eabc2d4044a0a71 : SUCCESS in 58m 58s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2561319668cd421880f331f8e8fb8503 : SUCCESS in 2h 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0710b38b7dac45aa8c42f186543fb853 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3ae6eb2d33d427f801cfd9c1192c9c8 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4f0769565f99715c1c97734de2c3afc95b3a3fdc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-03-22 13:34:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db4683f388f1fd178d32c5716786ffce6d99e9fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-22 16:03:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/606ca93c87db4ffda1901b16707fa5dc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8391850ddab482ca353616f2f9ac1e0 : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/724f6961d3a44eb9b6912042ba04f724 : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/be4ab3a3e99a4a008250602c906d93fd : SUCCESS in 42m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19b0ded0215c47168a7454484c07bb56 : SUCCESS in 1h 00m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbc6101cd5b34ebb8096368b132e5f55 : SUCCESS in 59m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10585a8160984aec90db86b215ed16ba : SUCCESS in 1h 02m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4447847f60d944cd96fdaa20f886da27 : SUCCESS in 1h 03m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc16b325a1c4466dab8a784737223bdf : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fc1b28e6c3e4e9ea9117acf21716934 : SUCCESS in 1h 43m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fef2ed14c2904e43a077bd768783c888 : SUCCESS in 2h 01m 44s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0606b0771da14a44bf774d340451be51 : SUCCESS in 2h 07m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97573c800d22494a9004838b1fc021c5 : SUCCESS in 1h 54m 49s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e54c0bca8e654044908bfbe597306d63 : SUCCESS in 1h 46m 11s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a9bc9088cac44dfbfebf0ce07df5485 : SUCCESS in 2h 02m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95fb56bcb3884d26b64d58e27bf6040a : SUCCESS in 1h 59m 33s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0bff66b669b4422ca007878d6a4f1413 : SUCCESS in 1h 04m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/527e07d3e49d42e39ab9a1541c8d68ee : SUCCESS in 1h 10m 56s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f092710a47b43508cff468f28a9441b : SUCCESS in 2h 13m 51s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/90a427f262de4071822ebf6a260e3ba8 : FAILURE in 46m 38s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d6b2ceaa48b4b17b0b842435d0116a6 : SUCCESS in 2h 17m 31s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbd5800ecc534e6fbd9b4cb734d84b04 : SUCCESS in 1h 08m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41a3168c0e147e684830c14bb88bd22 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/784bbd9e54784daba1a15320cd48b723 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e165d4419545909dbf7b8d03ca844c9e99a9ec3e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-03-26 10:44:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"78b8e6ddd1fa86b7bbc5c1b1a7d2d61fdc5aa5c1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-26 11:17:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"33efcdb4fe4272370b8f8db4f92b40b4d7c63def","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-03-31 08:48:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2f24f24290fa2240c17a1402244051d49f43b43a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-03-31 11:06:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b1e79eb8972d345e262705dad30e0f40bd2f9b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 11:08:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d9629e2c14af3657647849985c16936e20187da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 13:04:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abe725e810d446b98b4f1a5adb018fca\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e739f5f07b44ab4bc7803dec96b9532 : SUCCESS in 4m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6b09359643b9490ebef043a78a58b673 : SUCCESS in 39m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cffeb3deb8884b6880af4933bcf92a53 : SUCCESS in 38m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13eb1d2ce82f45d298c4abc3590d3ad3 : SUCCESS in 36m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d55930256fb44956b9e36a3646c17ffe : SUCCESS in 1h 10m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8b9b65a41c464d7a8c4e04d60efed352 : SUCCESS in 1h 05m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01cf864e48a546c580c138a28d7b6e87 : SUCCESS in 1h 04m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86cec4ec56684f47a3c3f77375c5b72b : SUCCESS in 1h 13m 05s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d145e7228584adbab663061492675ab : SUCCESS in 1h 55m 21s\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96054c0242114a869e8745f8b406dcd0 : SUCCESS in 1h 38m 57s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/371d7f77d1484d878b4a398af8a5a3f5 : SUCCESS in 1h 43m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a790c9c8054888a90de4a1f8128193 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08e2f194b97c4e83a4935a7070547b8e : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":3},{"id":"7497561c8e965104153f78b6612dd40e9142a536","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 13:04:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9a7f94ab7ce48de8a1194e1fd28d5078d986829e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 13:05:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8046fbfcdf4e414d89cfe1b50ccf94d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e62f68484a54a9a80b1a6e2db0a7198 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a76535685f5d64b7efb261dba04733f5e011ec28","revisions":{"1d66c7715c66e151c0a06b3aa06fdb730bb1bac4":{"kind":"REWORK","_number":1,"created":"2025-02-24 12:56:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/942581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/81/942581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/1"}}},"commit":{"parents":[{"commit":"4416d82c8927e61971e9bb50cfc171076633e2ab","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4416d82c8927e61971e9bb50cfc171076633e2ab"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-24 12:56:45.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-24 12:56:45.000000000","tz":60},"subject":"Align on cinder service naming","message":"Align on cinder service naming\n\nWith an effort [1][2] to align cinder service type naming to\n`block-storage` we revert deprecation of some variables, which\nwere annoinced as deprecated in [3] and dperecate versioned\nones instead.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/QQ7L44VE4MOWOQVOFCZQPTIVI35A6SCK/#X7BYL6AYQDMXAG7XXDNTSASWZCGEHQ37\n[2] https://review.opendev.org/q/topic:%22remove-cinder-v2-crud%22\n[3] https://review.opendev.org/c/openstack/openstack-ansible-os_cinder/+/934593\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/942580\nChange-Id: I7a8e563fa60a62bdc24d5a3c9d7fa1a43f324b76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/1d66c7715c66e151c0a06b3aa06fdb730bb1bac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/1d66c7715c66e151c0a06b3aa06fdb730bb1bac4"}]},"branch":"refs/heads/master"},"f2b467d9249396218569d5db5714947905212f37":{"kind":"REWORK","_number":2,"created":"2025-02-26 09:50:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/942581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/81/942581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/2"}}},"commit":{"parents":[{"commit":"4416d82c8927e61971e9bb50cfc171076633e2ab","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4416d82c8927e61971e9bb50cfc171076633e2ab"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-24 12:56:45.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-26 09:50:39.000000000","tz":0},"subject":"Align on cinder service naming","message":"Align on cinder service naming\n\nWith an effort [1][2] to align cinder service type naming to\n`block-storage` we revert deprecation of some variables, which\nwere annoinced as deprecated in [3] and dperecate versioned\nones instead.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/QQ7L44VE4MOWOQVOFCZQPTIVI35A6SCK/#X7BYL6AYQDMXAG7XXDNTSASWZCGEHQ37\n[2] https://review.opendev.org/q/topic:%22remove-cinder-v2-crud%22\n[3] https://review.opendev.org/c/openstack/openstack-ansible-os_cinder/+/934593\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/942580\nChange-Id: I7a8e563fa60a62bdc24d5a3c9d7fa1a43f324b76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f2b467d9249396218569d5db5714947905212f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f2b467d9249396218569d5db5714947905212f37"}]},"branch":"refs/heads/master"},"a76535685f5d64b7efb261dba04733f5e011ec28":{"kind":"REWORK","_number":3,"created":"2025-02-26 12:47:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/942581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/81/942581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/81/942581/3"}}},"commit":{"parents":[{"commit":"4416d82c8927e61971e9bb50cfc171076633e2ab","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4416d82c8927e61971e9bb50cfc171076633e2ab"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-24 12:56:45.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-26 12:47:02.000000000","tz":0},"subject":"Align on cinder service naming","message":"Align on cinder service naming\n\nWith an effort [1][2] to align cinder service type naming to\n`block-storage` we revert deprecation of some variables, which\nwere annoinced as deprecated in [3] and dperecate versioned\nones instead.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/QQ7L44VE4MOWOQVOFCZQPTIVI35A6SCK/#X7BYL6AYQDMXAG7XXDNTSASWZCGEHQ37\n[2] https://review.opendev.org/q/topic:%22remove-cinder-v2-crud%22\n[3] https://review.opendev.org/c/openstack/openstack-ansible-os_cinder/+/934593\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/942580\nChange-Id: I7a8e563fa60a62bdc24d5a3c9d7fa1a43f324b76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a76535685f5d64b7efb261dba04733f5e011ec28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a76535685f5d64b7efb261dba04733f5e011ec28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
