)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~949427","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I034b387fd8a3ded44be9b35b23ffdcd08eaf43d8","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa_hashi_vault","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-03-10 16:08:08.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-03-10 10:33:30.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-02-24 10:20:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-03-07 08:14:28.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I034b387fd8a3ded44be9b35b23ffdcd08eaf43d8","subject":"Add hashi_vault pki backend support","status":"MERGED","created":"2025-05-11 20:34:38.000000000","updated":"2026-03-10 16:09:37.000000000","submitted":"2026-03-10 16:08:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949427-osa_hashi_vault","meta_rev_id":"e8b1486a273de8042c24e432a16383b2ab4c31d2","_number":949427,"virtual_id_number":949427,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-10 16:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-02-24 10:20:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-07 19:22:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-03-07 08:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-03-07 08:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-11 23:08:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 14:34:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-02-24 15:17:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-02-24 15:17:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-02-24 15:17:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-03-07 19:22:26.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"57ff5d9272550a6aeebf956e4bcae71094fd7e2a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-11 20:34:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a76c9b3de1afb9d8601df68bf8934b3e1f090b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-11 23:08:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6594a73091043f0b3142963ebe272e4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7b1efc8a5374bf3ac8db59188120d60 : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a315dc67d74a40498dee912b447e40e6 : SUCCESS in 41m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2177cfa80e3f4c3e9847db119ea6bd06 : SUCCESS in 1h 20m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7f5827b49cff4d8a8fce797bcbd64b05 : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6a797804a04438eaaac750d332d0918 : SUCCESS in 45m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77f7295319d742a89d39a4a47844fe32 : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a108f328a6224a82a48dc2a81c8d0049 : SUCCESS in 1h 23m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad8e2e9b7b5147aebbcfb3ea818d7530 : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b48659648dc54b2d9d9ee72af93ea4ed : SUCCESS in 1h 25m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f0a7cd4acc849648cbb3188ddd068cd : SUCCESS in 1h 53m 49s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce1ddcdd520047da8e79f7bd0f8a6e93 : SUCCESS in 1h 54m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dae44245edca4e2da11ec77276f09a4e : SUCCESS in 1h 42m 41s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7ef1dbe31714f178358a7951bbe688c : SUCCESS in 1h 44m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c522103136774f71b94d8c2b4edc2d33 : SUCCESS in 1h 45m 43s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91c5adee95ee4ec988cde7e22a6307ad : SUCCESS in 1h 57m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4a1b9b3ebcee4b529d3c303410bdbdd1 : SUCCESS in 58m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f15004a98cc4f559a4c8214c2a66d59 : SUCCESS in 59m 10s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10971b2cced94da08a617a3e595243f0 : SUCCESS in 2h 18m 56s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ee6d8ed3b15493c98fb26c2995050a7 : SUCCESS in 1h 10m 21s\n- openstack-ansible-deploy-aio_ceph-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51f865e9604d4813bb0d12943ccb7044 : FAILURE in 43m 29s (non-voting)\n- openstack-ansible-deploy-aio_nfs-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18bc32a889314f9a9beaa8a9c4ffc484 : SUCCESS in 1h 50m 31s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a2e593c538849e4b0e44429430faa4a : SUCCESS in 2h 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c923c9a8c9ce4bc0bf661c253095e583 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"69d2afe4907833601a42538243159c31213e8b89","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-05-23 10:31:49.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"ce99da1b2e3dbf0dfd6ad298ab0169d5e7d69ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-10-07 19:38:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c1fb411d06a15c30446b393b3dfa2266863132eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 22:13:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4128c50f18fe41d59091fc8f088697ae\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44c6f34bd3fd4a7995f4e6d0e98d8abd : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/cb78abb28c8045b5802375d819a9141f : SUCCESS in 1h 02m 31s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/afbf301afe5844749b409e28a0afcb3b : SUCCESS in 1h 25m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/49c4a388da3042a68cff1ef75ad5ff07 : SUCCESS in 1h 05m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/74f74944ec5b4fc5a47e08fc417fcd7e : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10e633d9e1d747a9843b1fe54adc625a : SUCCESS in 1h 31m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56241ebe6aef475cbc2ce0cb6e019242 : SUCCESS in 43m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e75bda5e9df47e8a7e1a96323bbc0a7 : SUCCESS in 2h 13m 28s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33dde77edcaf46abb8c83e96ea115640 : SUCCESS in 2h 27m 56s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/94b18b3fab65497e9b2f3f09be9aad54 : FAILURE in 11m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b61113e7a7043ff9274f0e742c299a4 : FAILURE in 35m 07s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ceb6deb496c4d04a718026db9c2800f : SUCCESS in 2h 30m 26s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d50010b39c248d49938988b7277af8c : SUCCESS in 1h 49m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fddb85e9507a440aa8265395e5c529cc : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":2},{"id":"4d62a7cd33a2486a7b059d8746647af1d0cbf206","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-01-28 14:34:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7492bbdc5644a2e85b38b76a0fa9e1a207f7b6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-02-03 14:43:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e7d03c8b0bfd2ec8b28b8b52cf4f0706e3fde8d8","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-02-03 14:53:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0fbac04f4b12b22b1989485182371747a4a2292c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 17:04:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d812d4ef0a2847198219b6633e0cf8d6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af978d0f2eb94658aeabb5b443eca3e4 : SUCCESS in 10m 40s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7f1b1d78a1744b48ab3da4d892003110 : SUCCESS in 1h 10m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f16b3cee6ce1476c8b040246b3e8e1a3 : SUCCESS in 1h 25m 54s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/3e5490c3a1634aee8c6f735cacc1f088 : SUCCESS in 1h 31m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/20bf7911c7ee4827b753bce301d0e5af : SUCCESS in 53m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/deca34af85584f57b5d6ed931ca91bae : FAILURE in 1h 35m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec0a5f5a52d441a79e99bd3237e0943b : SUCCESS in 1h 32m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4bde0733f3e4cdca5389597b5f323b3 : SUCCESS in 1h 31m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d9cdc42e9ff64c828c2e39de2801583f : FAILURE in 1h 27m 14s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/372d02dc3efa417d9b4291d9db469e87 : FAILURE in 49m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b02babe1ae34f1cb38c1cda7b01bad5 : FAILURE in 42m 15s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a86d071a6c54190b692676191c674bd : FAILURE in 2h 00m 44s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e734ab3c3ab4aa5b9e11dc17459225c : FAILURE in 1h 22m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/221a9d439c464ee9a110f4a5a0698b14 : FAILURE in 9m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6393c5cfebcd4e3abf9b11a93279e485 : FAILURE in 57m 47s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbc5c5e01c714785b3944a3281d4da12 : SUCCESS in 2h 02m 02s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d618fb9e87a4d789287e9d0b974e3ea : SUCCESS in 1h 06m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280c8484b01a473e973f60fde745e4d5 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":3},{"id":"32218f4fcd1b97a607113984edb7b958d37c902f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-02-05 17:03:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"574dcc42992c53722e67d74f638490ac37968d3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 19:42:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b68721254e434e97aafa854b536652a0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c41d5598f714e889aa0a96de42c222d : SUCCESS in 16m 46s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/048da4d1c4d84429b31d934a96434250 : SUCCESS in 43m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a0af1c3fd3a3470692ea2ac9cbe0203b : SUCCESS in 1h 34m 47s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7d29dc6145d84eea8b9d3b195b901ffa : SUCCESS in 1h 30m 48s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e5cc459e3c6f47fda333b04c44244a43 : SUCCESS in 50m 19s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/54cc4d7225af4938a582201ce54b1e62 : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22441896260c40e2874f37290ddb5adf : SUCCESS in 1h 29m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1c8d09cc7454cd691945841ee1a2aff : SUCCESS in 1h 26m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b5b3cf07017343cb8a255ad0090e4435 : SUCCESS in 1h 42m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4ab3692fdb484419a571e8940bb8979d : SUCCESS in 1h 09m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/211b262b013c423aabe4d46abdce84f8 : SUCCESS in 2h 15m 20s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcea72d8baa14d95b01219e56649cbc4 : SUCCESS in 2h 16m 38s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26eeff0c7c234658802a7d24e2eb690e : SUCCESS in 2h 29m 42s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3e3fa5b8e00344cb8b07ff8be2f4a751 : FAILURE in 18m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bacb86f75284d9083f8253c5f08aa5b : FAILURE in 32m 33s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80d28a35619a4024971274a8776352f1 : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93b7e0b8e0154129933ca68f95183c33 : SUCCESS in 1h 07m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ecb0177bba94aee9c9ebf832bccb173 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ed1fee3e32e3a345bad8cdc207e2bc6675e86b78","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-02-24 10:20:39.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fcee85c5e8778e47c39b856a0c91f9e72e9d0da","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-24 15:17:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"71554181157729812f26ecb52bf1d5c4fae91dbc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-07 08:14:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8918f71ffea92fd7f6324f00c83f98fac95c3383","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 08:16:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"83ddd1872edd937b0af0f833a5b61c4a28b0db69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 10:40:51.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05e0415317aa4495b1f6e745509f9094\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16551e1f2b214040a98fcb9f7b6884ea : SUCCESS in 9m 06s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b2b75064d8624c0e89b9f1eca0aba8c2 : SUCCESS in 1h 29m 44s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/035873e7030c4131b851ac28e5f39310 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a4e55fa231df4e3789195773814c3d6d : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6f63734d9dcb4ba8b51db9128b4ad0ce : SUCCESS in 47m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b5e5f8dae4b9463a92092c21e1341b50 : SUCCESS in 1h 39m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d82f4b45f0a425093e9456f48511436 : SUCCESS in 1h 28m 10s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d211c0e840724de9bbe086854ee3e221 : SUCCESS in 1h 38m 24s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83f632763e8f420b9dd820749938982a : FAILURE in 1h 58m 07s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ccf0d1b44c54f65a15c4a6fa1df8eb8 : SUCCESS in 2h 14m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b713576a89374e6e9494325527bd8af0 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a19e5f43f39b6bfa79eefd1ac5036323d9d04ae2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-07 19:22:26.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9279ba2527117a18b7ac114884c473da7322693a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 21:54:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/208e2e82eb2240f189515ca9e16078cc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dce75c218fa741a6b3e327d100a762d0 : SUCCESS in 17m 46s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9752b089930348fdbe20cba22c9ca4ee : SUCCESS in 1h 05m 45s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/15e16c35edd64aba8bd42fa7a9d4746c : SUCCESS in 50m 44s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7f0c11ebc63344e5934cfc2660e81b51 : SUCCESS in 1h 44m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6dcffd655f224c91b2db69a4ab04848e : SUCCESS in 1h 39m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b446d390661a4876b35e8873ec21e591 : SUCCESS in 1h 40m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90ebcfe1e79e459c9b671dddf96b80a9 : SUCCESS in 45m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ee3334b081041eb9d0dcb91f22e1265 : FAILURE in 1h 18m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c968a2025c0c445ba24c2f0f7e446440 : SUCCESS in 1h 02m 16s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/00003d15781448b2936ddc9b266aa246 : SUCCESS in 1h 57m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/525f5791dc284b82824daeeb4bcee4b2 : SUCCESS in 2h 24m 44s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae081952899e428c9e1cdf4f8ef3c270 : SUCCESS in 2h 21m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b072444687554befbead2e47186ba1df : SUCCESS in 1h 04m 17s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/02ec6b22ed2a437681f467dc3e8eb055 : FAILURE in 13m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70c5eb1d61cd480ebc27f1e72e003ea4 : FAILURE in 14m 49s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32117e9e2d6f4280875279d3409470f9 : POST_FAILURE in 1h 22m 20s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8c6fa96df514cda84445e242a6e0eab : SUCCESS in 2h 17m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0ab412f9e14d7890eecc75f7b1c2fd : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":3},{"id":"0b575eb91e42f543fd1911edc1d2d88406e27f69","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-09 22:36:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c734d27634a3b9345c9b38d39b5e153fccc67a6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 00:54:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6af32e6caba4b9d998c719c2b5578ed\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba2ee38b4bf14a909bbadff4d5591c1d : SUCCESS in 9m 32s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b4812e5d38a64efd98ab1df0d5b1e672 : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2e2d3f3d527a46aab571f8f7dc96d5ed : SUCCESS in 1h 31m 25s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/85f2d1a88a9d4b80a175fb3d9b27dd62 : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2bcce0a5a5f74a1b9d1bfe888f6970c2 : SUCCESS in 1h 08m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c281a6038e4c49ceb8dfbb444014b468 : SUCCESS in 47m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fcaf39d986e4372980f164e8e4da39d : SUCCESS in 1h 32m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0083c9ac7ba4b14b3441eed202f236a : SUCCESS in 54m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5fe89c380d234322890e8befe96bc93d : SUCCESS in 1h 38m 59s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3fd89bc63a5649f589f8071ba0667f2a : SUCCESS in 1h 08m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b34bf95b9f724ea5a230625ab2e742e1 : SUCCESS in 1h 04m 50s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/768b24c88817494c8164208c11e1ecbf : SUCCESS in 1h 01m 07s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d98cda188324893bc1888529f0ddc80 : FAILURE in 1h 08m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/33acb5a0a9a04059a3b93005c7b770f7 : FAILURE in 9m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75bceb1b13e0467192cdc9217aea43be : FAILURE in 17m 17s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d6a9ce17bd24860af5df46d234bc381 : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9284816abf54ef89a7d5c0493a1ff28 : SUCCESS in 2h 15m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de401918c7ff42efb50aea9addeb1b1c : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6dcb07551d381b605b698c0a46f441aef7668b50","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-10 10:33:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"427cff1534cddb9d0b68f5a32bf2a69bc8fe8db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 13:25:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fbf862fa45f437d8d3bb1d8d6c14fe4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84de7550621d4249becd6f1609a191b1 : SUCCESS in 16m 21s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d2fc8de410334333bdd47f9f34af8db3 : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f762fb11232a4e8aad1f8b4919e7934b : SUCCESS in 48m 04s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c651ab030f2847d89b1123c677bb0b93 : SUCCESS in 1h 36m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/feb8c38623c246e4a4e918281538d922 : SUCCESS in 57m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b784fc65696c4131ae33917abe43a427 : SUCCESS in 50m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/336c82956ffa468993294b4b00beb9e7 : SUCCESS in 1h 30m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ba4eaaeeefe4d0d9b46ca54913e9899 : SUCCESS in 49m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5e7140df1f6b4012b0a5680524f0ea44 : SUCCESS in 2h 14m 19s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/96e99565423b48dca3bd084bf2efc5cc : SUCCESS in 2h 16m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6b997b27d3f49b49a977b65d3ce9877 : SUCCESS in 2h 17m 58s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8052ad6a55c4b9695b9fd32dbbabc39 : SUCCESS in 2h 17m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5043b3e23b541f2a58d0bf7c989b4e7 : SUCCESS in 1h 54m 31s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eb90de2050354723b540f534c0440d26 : FAILURE in 8m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea9044ea38c8408db05c3c5ea15ce74f : FAILURE in 30m 17s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6805acec4c7d4a438466181c449b370d : SUCCESS in 2h 22m 40s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1ec7268edde4fdb8c7006f0b93f8355 : SUCCESS in 2h 33m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3509c0e08aa84daa913efcf655ff85f9 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"1cf5a41b76d10e2537218175f91c47bf836db089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 13:27:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b25552eb1a7cbc0095a944f9317c48065f726db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 16:08:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f46e5068fb424ca2185de5848ce921\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fbcba1c90494078bfc375d4b3cf8a10 : SUCCESS in 14m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/4a1a704001254728b2467a22f896f7db : SUCCESS in 1h 27m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/901a4fe749da4a2eabf16a4d6b9b3ac3 : SUCCESS in 1h 29m 57s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bdae44d0807f4c77b15b1299f237c14e : SUCCESS in 1h 27m 29s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4c6be74657b8423ea5b3e98693e34f16 : SUCCESS in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4030bdf9a8fe43dca8101eab38072da5 : SUCCESS in 1h 43m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a609a90d64a948d996cc761d7f1938b9 : SUCCESS in 1h 37m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4f2180000434402a4122276936140ba : SUCCESS in 1h 41m 24s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0b7c8428dba48df95e7e37d7e6fb3b3 : SUCCESS in 2h 06m 13s\n- openstack-ansible-deploy-aio_nfs-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c88056629504d73ac7245b6dbb6424b : SUCCESS in 2h 35m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda2d9f6354044fbace9d88570a2e962 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":3},{"id":"078a5ac2d52704ad0d1ce53c99f43536d4f54c09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 16:08:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e8b1486a273de8042c24e432a16383b2ab4c31d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 16:09:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c6ed33510e4bad9f1606a4d332cfef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b04b0407f52429db75e98d0469b3702 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c5759838757b795c28756771d67b64510f5e3179","revisions":{"95d9ae6e89c8df3b6fad7d40040428a6cf7a7d38":{"kind":"REWORK","_number":1,"created":"2025-05-11 20:34:38.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/27/949427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/27/949427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/1"}}},"commit":{"parents":[{"commit":"4ee2eb4e9b9caa3267c948e3dcb5e3b86a46646c","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4ee2eb4e9b9caa3267c948e3dcb5e3b86a46646c"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-08 16:31:14.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-11 20:34:36.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `cinder_pki_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n\nChange-Id: I034b387fd8a3ded44be9b35b23ffdcd08eaf43d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/95d9ae6e89c8df3b6fad7d40040428a6cf7a7d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/95d9ae6e89c8df3b6fad7d40040428a6cf7a7d38"}]},"branch":"refs/heads/master"},"892d20d08524dfe20b92e6676f48529d4301bbcc":{"kind":"REWORK","_number":2,"created":"2025-10-07 19:38:55.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/27/949427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/27/949427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/2"}}},"commit":{"parents":[{"commit":"4ee2eb4e9b9caa3267c948e3dcb5e3b86a46646c","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4ee2eb4e9b9caa3267c948e3dcb5e3b86a46646c"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-08 16:31:14.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-10-07 19:38:37.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `cinder_pki_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nChange-Id: I034b387fd8a3ded44be9b35b23ffdcd08eaf43d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/892d20d08524dfe20b92e6676f48529d4301bbcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/892d20d08524dfe20b92e6676f48529d4301bbcc"}]},"branch":"refs/heads/master"},"c5759838757b795c28756771d67b64510f5e3179":{"kind":"REWORK","_number":3,"created":"2026-02-03 14:43:29.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/27/949427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/27/949427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/27/949427/3"}}},"commit":{"parents":[{"commit":"4ee2eb4e9b9caa3267c948e3dcb5e3b86a46646c","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4ee2eb4e9b9caa3267c948e3dcb5e3b86a46646c"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2025-05-08 16:31:14.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-02-03 14:43:19.000000000","tz":60},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `cinder_pki_san` parameter to the new one\nchanged in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\nChange-Id: I034b387fd8a3ded44be9b35b23ffdcd08eaf43d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/c5759838757b795c28756771d67b64510f5e3179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/c5759838757b795c28756771d67b64510f5e3179"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
