)]}'
{"id":"openstack%2Fopenstack-ansible-os_cloudkitty~1003640","triplet_id":"openstack%2Fopenstack-ansible-os_cloudkitty~master~I8bfa2b391cd29a96eb5ba0394f8a4e5dfc605e92","project":"openstack/openstack-ansible-os_cloudkitty","branch":"master","topic":"osa/pep-508","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-03 16:56:12.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8bfa2b391cd29a96eb5ba0394f8a4e5dfc605e92","subject":"Use Direct URL syntax for pip packages","status":"NEW","created":"2026-09-03 10:03:00.000000000","updated":"2026-10-05 19:34:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0256911a924e3653363232695f08920c5aa211eb","_number":1003640,"virtual_id_number":1003640,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-03 16:56:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 16:56:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 10:54:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2026-09-25 02:55:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"5420fa17bbcd231b190496fba6ab2560764a9f7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-03 10:03:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a612f9f1bbddaaa109b36d5871af31a2d604624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:56:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15f9465e51224135997ca6c61df4e200\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1529712e9a04a23aa9ce9f659e6d05b : SUCCESS in 13m 24s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/59213d1895094c1a984a0c6c6497b25d : SUCCESS in 1h 38m 29s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9a00b7a20bc74d0faf8c8b9941a3e106 : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b67d23e0ad934c6594119c7630092a8c : SUCCESS in 1h 29m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9b1aedd2292a495c96184237c8a96f01 : FAILURE in 18m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64102836462647eab8e8caf89d75e355 : SUCCESS in 1h 50m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/bc942b8e9c4646ce993cb38564268d4a : FAILURE in 1h 45m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b5d59fe90d6497580bddbe73d8d9ca8 : SUCCESS in 1h 52m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d4e33ace627b41eaa2b1df7186949c05 : SUCCESS in 2h 13m 02s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15df213ed9a24bd1982a1d3b12e7bb0a : SUCCESS in 2h 51m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc6a43ef83a3450e87519b7cdc9c85e1 : SUCCESS in 2h 41m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e8b38b83364f9698a576e5aadd0e8d : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02baf56600634ef49b6405dd9ccf2323 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"54477981896c0cbef10606ad0fc5cef771f7dcef","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-09-18 10:54:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c4fa1fd94ba2db2c6e298d2aee5cb4d83d076f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 14:17:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4209109c549449a396661924aa57f98d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c1d5b6354a845d6854e75a336adb284 : SUCCESS in 14m 59s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/58742e5d6e164058ad8faa7d603347b9 : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a2c15837d4ca4b78902d287bc1a4ce8d : SUCCESS in 1h 47m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ab086e4357d34414bc0a3a5165b86f06 : SUCCESS in 1h 35m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7a7c8d91fe354b1188a1db40851909b0 : SUCCESS in 1h 32m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70228d0510d94bce956cbba2b4c6ca55 : SUCCESS in 1h 41m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/46f2cecd26fe4615ad84f4096bee371e : FAILURE in 1h 40m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b929f2c6fd764e218d31188e925922d0 : SUCCESS in 1h 47m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/99affc7e15f14db18d3980ab88a71d9d : SUCCESS in 2h 48m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15e4a18d75ab4337a47edd629a60cb87 : SUCCESS in 2h 48m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56648885351548db9e7e897459fb1c3a : SUCCESS in 2h 43m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3be303039a84dba9aa5bafbbd3ae906 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd97a70563b34ebfb2606774f050b15a : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e0084235f1d8011a0aaa4cd2101e6d8d256ca488","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-25 02:55:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9af342d98a9504e1c650ea2518115d90785fffbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:25:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb61dfe63b96466ab09dbf5c02f9ea49\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89fb0c48367744c5b60b98615d20cf02 : SUCCESS in 17m 06s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5c44f5e1c4714bceb24bd9a55e0a6bfa : SUCCESS in 1h 14m 51s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f6527cba3cb040a3a96a087e94d1c63f : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/79445696d0594193bd2d8baa44d770fa : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d8758b677183406fa2488cbb29c92b9f : SUCCESS in 1h 35m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57acb8a84589451c8eea039a92e9312d : SUCCESS in 1h 38m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5e77caf0a3cb4194b765ce3da6e76692 : FAILURE in 1h 37m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39c0fc75a987467fa15dd4ef8d974750 : SUCCESS in 1h 55m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ef62c1afa4cb45efa56d4ed373c2389d : SUCCESS in 2h 30m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d35f3b7b3d614deaa95b55e580ccd2b1 : SUCCESS in 2h 39m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e84430be9fc54651b03638103e06c39d : SUCCESS in 2h 56m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0d25fbce854b8a950d3e2d4efbc4a5 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c29bfb097cd4ceeb305b0751073a400 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d3198679e2bd0f5996666cc3d7de4789001fb81b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-29 20:19:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b647f3ba34eca30d91abe47445a24eeecc77eafb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 23:13:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c997ae949154e7d870415e8d5301382\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c0b97bacce04689b9bf3b8407ac9e15 : SUCCESS in 15m 43s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/706ef5217d594b118615e093b270b0a4 : SUCCESS in 56m 57s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/303609c5ac4448b090413fad59c285a5 : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9bff403209dd494dbf1e979d749aa787 : SUCCESS in 1h 49m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/28f928897dab43e6bac62657358605f0 : SUCCESS in 1h 27m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0df241afd9134ebb86839c78a703326e : SUCCESS in 1h 45m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/553a6ef9673f47c2983d69ee27e8da66 : FAILURE in 1h 16m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fe4e28846834547bf38a995b50fc457 : SUCCESS in 1h 30m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8b5b6ed5a2cd448f8f13ba8efdc3a297 : SUCCESS in 1h 38m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75af2e671db2498db2e5e2bfe8ac8530 : SUCCESS in 1h 35m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac68eba5528f4b6ba37bb80e91b7c289 : SUCCESS in 2h 43m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e52ea077597b4c3eac391f23be0c07a3 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/400a027f8370440e8eac0a40923649c6 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ae0b3dfaec1c08c13677915c5eca5bf75d167b97","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-05 09:31:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"989047c5a6e158a7badc432274f21e271238f698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 12:16:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2221acc993664c08a9b8c0febac07f1e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22ae702339b74c60b8d579767d101e92 : SUCCESS in 9m 41s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1e9717a4ad8b436bbbe766ccf643a646 : SUCCESS in 48m 04s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/99727e22764e421387152f2e0d507ae1 : SUCCESS in 1h 26m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/29ee3b6a216643d6a92ae7689e8ebd29 : SUCCESS in 1h 35m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae91dce928194c86a73f29d40b281af7 : SUCCESS in 54m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25682ac131754f3faa8142466aaa3216 : SUCCESS in 1h 35m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/40730cf1e6e44775bdd66a7fd23dbdad : FAILURE in 1h 24m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bf92ae21336416e9c09196a0a667584 : SUCCESS in 1h 36m 22s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8415a3d78b88448da2c51634cff51565 : SUCCESS in 2h 37m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d84a73e5198b42ae894cac7e7c2d1ed2 : SUCCESS in 1h 18m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78584144b33b44edab8730e2489dce87 : SUCCESS in 1h 21m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad754c816204eee8ed4a82b44aba253 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6f4a953d6b14f04bee4bc295ed14889 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0e75267ac1359e6ef66fd3c018fa112d880ec50b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-05 17:17:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0256911a924e3653363232695f08920c5aa211eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 19:34:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1452a576b42849b68d8820114f613db7\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47dc7d274dbd4515aec6c385dd596b30 : SUCCESS in 11m 24s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9d8219dc23094fc5877da57f839d146e : SUCCESS in 1h 43m 57s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f468b490486f42c3b38055616643722d : SUCCESS in 1h 38m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/78d6b0ab1a9f4311b76496a8e625aa2d : NODE_FAILURE Node(set) request 49a7e29146bb40f2943222bd47a71408 failed in 0s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5cfec5b8275f41709990689ca0931627 : NODE_FAILURE Node(set) request 6127f143ae2348958be318f9f4ca3175 failed in 0s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9147e311f204648916f5878fd25a12d : SUCCESS in 1h 50m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ad28835936b84e5c8c8952bc89434137 : FAILURE in 1h 29m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32719027844d4cd781dbc27bfbc7ae86 : SUCCESS in 1h 44m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/156bb30de1d74f54b2812cda5ce07486 : NODE_FAILURE Node(set) request be1c8ecfa11a4657920805150b9712e4 failed in 0s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a424324f26a4ad4b12c328af02ccfbe : SUCCESS in 1h 54m 15s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7621a433dff24d9aa7237f77917f24d9 : SUCCESS in 1h 55m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ad113d9f67420b9782ddf0cdeff8f1 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15fedd283d5d46d4bef0f6aa659d55bb : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7bb58eddfe865ffdca51677ebf3268b004469f0","revisions":{"e7bb58eddfe865ffdca51677ebf3268b004469f0":{"kind":"REWORK","_number":1,"created":"2026-09-03 10:03:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/40/1003640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty","ref":"refs/changes/40/1003640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/40/1003640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/40/1003640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/40/1003640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/40/1003640/1"}}},"commit":{"parents":[{"commit":"152bad87e9bed3ba9f75c110101c1a12dcb8fc18","subject":"Use correct libxml package name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/152bad87e9bed3ba9f75c110101c1a12dcb8fc18"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-03 10:02:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-03 10:02:51.000000000","tz":120},"subject":"Use Direct URL syntax for pip packages","message":"Use Direct URL syntax for pip packages\n\nUsage of #egg has been deprecated and supplying extra requirements\nfor the egg is no longer supported since pip 26.0 [1]\n\nUsage of Direct URL is conformant with PEP 508 [2]\n\n[1] https://pip.pypa.io/en/latest/news/#v26-0\n[2] https://peps.python.org/pep-0508/\n\nChange-Id: I8bfa2b391cd29a96eb5ba0394f8a4e5dfc605e92\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/e7bb58eddfe865ffdca51677ebf3268b004469f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/e7bb58eddfe865ffdca51677ebf3268b004469f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
