)]}'
{"id":"openstack%2Fopenstack-ansible-os_cloudkitty~429629","triplet_id":"openstack%2Fopenstack-ansible-os_cloudkitty~master~Ie6a1ec3cad1ff192535d054c0ce3cf58800ab651","project":"openstack/openstack-ansible-os_cloudkitty","branch":"master","topic":"use_https","hashtags":[],"change_id":"Ie6a1ec3cad1ff192535d054c0ce3cf58800ab651","subject":"Use https instead of http for git.openstack.org","status":"MERGED","created":"2017-02-06 10:21:43.000000000","updated":"2017-02-06 10:27:33.000000000","submitted":"2017-02-06 10:27:33.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"facb6f1791e4911a8da47fbfbbf1708801b3b7e4","_number":429629,"virtual_id_number":429629,"owner":{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2017-02-06 10:27:33.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2017-02-06 10:25:12.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2017-02-06 10:25:12.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-02-06 10:25:12.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2017-02-06 10:27:33.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"97921080d994aa122ca2f726c09492129d35e8d9","author":{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},"date":"2017-02-06 10:21:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2074f14f42f2c3ad39ae382bc615b48aa0d9bb38","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2017-02-06 10:25:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3bc51c73ac89fb1fe5a22a16af575df793d26f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-06 10:25:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-os_cloudkitty-docs-ubuntu-xenial http://docs-draft.openstack.org/29/429629/1/check/gate-openstack-ansible-os_cloudkitty-docs-ubuntu-xenial/798daa2//doc/build/html/ : SUCCESS in 1m 16s\n- gate-openstack-ansible-os_cloudkitty-linters-ubuntu-xenial http://logs.openstack.org/29/429629/1/check/gate-openstack-ansible-os_cloudkitty-linters-ubuntu-xenial/c238284/ : SUCCESS in 1m 45s\n- gate-openstack-ansible-os_cloudkitty-releasenotes http://docs-draft.openstack.org/29/429629/1/check/gate-openstack-ansible-os_cloudkitty-releasenotes/665f205//releasenotes/build/html/ : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"6872959c934c36c4e73425a9b85f3f8663e3af06","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-06 10:26:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"77f9d4dd1d303f92fd718d8d76bed555abe1eabe","date":"2017-02-06 10:27:33.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"f97e59d0a5ccadf149c7c3ab5e3cd72c897646bd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-06 10:27:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-ansible-os_cloudkitty-docs-ubuntu-xenial http://docs-draft.openstack.org/29/429629/1/gate/gate-openstack-ansible-os_cloudkitty-docs-ubuntu-xenial/ce48289//doc/build/html/ : SUCCESS in 1m 09s\n- gate-openstack-ansible-os_cloudkitty-linters-ubuntu-xenial http://logs.openstack.org/29/429629/1/gate/gate-openstack-ansible-os_cloudkitty-linters-ubuntu-xenial/1ba2486/ : SUCCESS in 1m 16s\n- gate-openstack-ansible-os_cloudkitty-releasenotes http://docs-draft.openstack.org/29/429629/1/gate/gate-openstack-ansible-os_cloudkitty-releasenotes/610e718//releasenotes/build/html/ : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48412b6941c9291cf1c8b309e528dc16adf7bddd","revisions":{"48412b6941c9291cf1c8b309e528dc16adf7bddd":{"kind":"REWORK","_number":1,"created":"2017-02-06 10:21:43.000000000","uploader":{"_account_id":15905,"name":"Cao Xuan Hoang","email":"xuanhoangqb@gmail.com","username":"hoangcx"},"ref":"refs/changes/29/429629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty","ref":"refs/changes/29/429629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/29/429629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/29/429629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/29/429629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/29/429629/1"}}},"commit":{"parents":[{"commit":"3b6300f3ce889b04d4e83cea93a2bcb0f499b06c","subject":"Phase out trusty and use systemd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/3b6300f3ce889b04d4e83cea93a2bcb0f499b06c"}]}],"author":{"name":"Cao Xuan Hoang","email":"hoangcx@vn.fujitsu.com","date":"2017-02-06 10:19:44.000000000","tz":420},"committer":{"name":"Cao Xuan Hoang","email":"hoangcx@vn.fujitsu.com","date":"2017-02-06 10:20:18.000000000","tz":420},"subject":"Use https instead of http for git.openstack.org","message":"Use https instead of http for git.openstack.org\n\nTrivialfix\n\nChange-Id: Ie6a1ec3cad1ff192535d054c0ce3cf58800ab651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/48412b6941c9291cf1c8b309e528dc16adf7bddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/48412b6941c9291cf1c8b309e528dc16adf7bddd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3b6300f3ce889b04d4e83cea93a2bcb0f499b06c","is_merged_in_target_branch":true,"change_id":"I50d377ce6c5e80386954fd551e566c55f57c5a3a","change_number":423141,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
