)]}'
{"id":"openstack%2Fopenstack-ansible-os_cloudkitty~758743","triplet_id":"openstack%2Fopenstack-ansible-os_cloudkitty~master~I6a73c0631aaf46835c3d0fce0bf810509a30bb17","project":"openstack/openstack-ansible-os_cloudkitty","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I6a73c0631aaf46835c3d0fce0bf810509a30bb17","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-10-19 09:19:48.000000000","updated":"2020-10-19 12:54:49.000000000","submitted":"2020-10-19 12:53:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758743-1603112012212-7a5aba1f","meta_rev_id":"7d5718bd518b91ecc3e19ff5dd502ca9236c5906","_number":758743,"virtual_id_number":758743,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 12:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-19 12:11:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-19 12:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 12:11:24.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-19 12:53:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c08eb258ce758308f20b8776e5cd195e5ba1551","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-19 09:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fa8fe612e9ff9e20ebc7bd85b5ca5d3c7053da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 11:16:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a1e03efadb7748e59865a73f5ed2d8e6 : SUCCESS in 11m 22s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/53e0762b007347069927db8135a212b6 : SUCCESS in 33m 59s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d1f5353e99d641368f67f9ba51466419 : SUCCESS in 37m 16s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fd9bdc3787f445bb544824339ce3327 : SUCCESS in 34m 22s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c23897d52fa945f696fbc51f47012872 : SUCCESS in 31m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7601735396a4d9f906b67acd0106b7a : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":1},{"id":"497f642e1c6c046e2fc84d7c5f17a102102bfe4c","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-19 12:11:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"745cb5280d88f950b6456877b7f1d913e22b4643","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 12:11:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81cb174f1b2191bf46d639422cdf18348c8b17d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 12:53:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5cd88f97466d4d4580d7ca20517dca59 : SUCCESS in 11m 37s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/41e70765011f494daf1810b408e4c559 : SUCCESS in 31m 39s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b83f21e1f414fc5b1a05f0f5dcfd3b4 : SUCCESS in 37m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1627d953f12c4c9799e626bf19ee7bc6 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f70bc1cb2f3467af70eeb9f2d7857945db84e8f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 12:53:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b62df8fec5f9f92565346ab0cf62df13b244e809","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 12:54:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbdcd1ec01fa43d192d875916b423b89 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ac496f43eb58d3e47ef469e9bb651d7685bff9d","revisions":{"7ac496f43eb58d3e47ef469e9bb651d7685bff9d":{"kind":"REWORK","_number":1,"created":"2020-10-19 09:19:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/758743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty","ref":"refs/changes/43/758743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/43/758743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/43/758743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/43/758743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cloudkitty refs/changes/43/758743/1"}}},"commit":{"parents":[{"commit":"bd7220d8ca1ebab6e1a27229b47c917e2d2caa4a","subject":"Merge \"Define condition for the first play host one time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/bd7220d8ca1ebab6e1a27229b47c917e2d2caa4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:19:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:19:47.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I6a73c0631aaf46835c3d0fce0bf810509a30bb17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/7ac496f43eb58d3e47ef469e9bb651d7685bff9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cloudkitty/commit/7ac496f43eb58d3e47ef469e9bb651d7685bff9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
