)]}'
{"id":"openstack%2Fopenstack-ansible-os_congress~613262","triplet_id":"openstack%2Fopenstack-ansible-os_congress~master~I102c4540efb87aeb26bf3f26f7725a95021a1748","project":"openstack/openstack-ansible-os_congress","branch":"master","topic":"update_doc","hashtags":[],"change_id":"I102c4540efb87aeb26bf3f26f7725a95021a1748","subject":"add bug url to readme","status":"MERGED","created":"2018-10-25 10:45:03.000000000","updated":"2019-04-24 19:44:53.000000000","submitted":"2019-04-24 19:44:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"613262-1556135093422-7a5aba1f","meta_rev_id":"07462e1fddbacf90692fdeb497ec04ab5633913b","_number":613262,"virtual_id_number":613262,"owner":{"_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},{"value":0,"date":"2019-01-10 11:55:17.000000000","_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 19:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"recommended":{"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},"all":[{"value":1,"date":"2018-10-31 01:15:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},{"value":0,"_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"},{"value":2,"date":"2019-04-24 19:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},{"value":0,"_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"},{"value":1,"date":"2019-04-24 19:10:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},{"value":0,"_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},{"_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-31 01:15:49.000000000","updated_by":{"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},"reviewer":{"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},"state":"REVIEWER"},{"updated":"2019-04-24 19:10:12.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 19:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c93db1bf3ecc39d6ca7d1384b0c36fd1dc8cc83b","author":{"_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"},"date":"2018-10-25 10:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6590e88b10112853ea5122489882d0b66dff7ed6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 11:45:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/62/613262/1/check/openstack-tox-docs/181f5fd/html/ : SUCCESS in 3m 01s\n- openstack-ansible-linters http://logs.openstack.org/62/613262/1/check/openstack-ansible-linters/1f985de/ : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e0f73597e0511badd47b6ee42a3cdb5ff7ffafca","author":{"_account_id":26775,"name":"wanghui","email":"wang_hui@inspur.com","username":"wanghui"},"date":"2018-10-31 01:15:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3ea08e399585fea9ac385cedd85b699b7c2b2cc2","author":{"_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"},"date":"2019-01-10 11:55:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f38a17f0f7aa23e73849f92a22b7d06f6e1ecde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-10 12:05:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/62/613262/1/check/openstack-tox-docs/0387d05/html/ : SUCCESS in 3m 30s\n- openstack-ansible-linters http://logs.openstack.org/62/613262/1/check/openstack-ansible-linters/c4abaa3/ : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b269d667071106dcfca4ccdf42533dbf4672d387","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 19:10:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3c1b01b0c4424696890c63b890b158a6ae2def1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:13:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f4225354815b214aac7d82877604cb536b6646f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:44:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c56907de9329f9605652cd1273e047468f482039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:44:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/62/613262/1/gate/openstack-ansible-linters/cf79512/ : SUCCESS in 10m 51s\n- openstack-tox-docs http://logs.openstack.org/62/613262/1/gate/openstack-tox-docs/eae0793/html/ : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a43556f853431e0446f5779607401a73ee384f13","revisions":{"a43556f853431e0446f5779607401a73ee384f13":{"kind":"REWORK","_number":1,"created":"2018-10-25 10:45:03.000000000","uploader":{"_account_id":27385,"name":"lvxianguo","email":"lvxianguo@inspur.com","username":"lvxianguo"},"ref":"refs/changes/62/613262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_congress","ref":"refs/changes/62/613262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/62/613262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/62/613262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/62/613262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/62/613262/1"}}},"commit":{"parents":[{"commit":"934727a3e05db6dacb656a461fe9be9054c3a753","subject":"Set container_name as host_var in test inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/934727a3e05db6dacb656a461fe9be9054c3a753"}]}],"author":{"name":"lvxianguo","email":"lvxianguo@inspur.com","date":"2018-10-25 10:43:16.000000000","tz":480},"committer":{"name":"lvxianguo","email":"lvxianguo@inspur.com","date":"2018-10-25 10:44:46.000000000","tz":480},"subject":"add bug url to readme","message":"add bug url to readme\n\nChange-Id: I102c4540efb87aeb26bf3f26f7725a95021a1748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/a43556f853431e0446f5779607401a73ee384f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/a43556f853431e0446f5779607401a73ee384f13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
