)]}'
{"id":"openstack%2Fopenstack-ansible-os_congress~691968","triplet_id":"openstack%2Fopenstack-ansible-os_congress~stable%2Fstein~Ia6e1341dd93a2cbf78179d1064b924b4d3126470","project":"openstack/openstack-ansible-os_congress","branch":"stable/stein","topic":"use-opensuse-15","hashtags":[],"change_id":"Ia6e1341dd93a2cbf78179d1064b924b4d3126470","subject":"Switch to opensuse-15","status":"MERGED","created":"2019-10-29 20:36:35.000000000","updated":"2019-10-30 14:50:09.000000000","submitted":"2019-10-30 14:46:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691968-1572446793087-df6c426c","meta_rev_id":"5e1c5130e262c7f531d6a0ddcdb688f9f8a7060b","_number":691968,"virtual_id_number":691968,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 14:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-10-30 12:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-30 13:36:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-30 13:36:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 21:56:26.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-10-30 12:35:20.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-30 13:36:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-30 14:46:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"892a9306a826bdcf4aff4d81a107656636e505fe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-29 20:36:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f7a7f2ad80e46f6e92004d40439e5a1239b1049","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-29 20:42:41.000000000","message":"Topic changed from use-opensuse-150 to use-opensuse-15","accounts_in_message":[],"_revision_number":1},{"id":"8b68bbf660400bd52af1ebd0c38561837a8ae588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 21:43:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a5d68f8babad427e97d9d57833fd3c16 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0bf7740b2543f2b80eedb5ab4b1e9b : SUCCESS in 3m 06s\n- openstack-ansible-deploy-aio_metal_congress-debian-stable https://zuul.opendev.org/t/openstack/build/28888d5763aa477cb74075218c2912a5 : SUCCESS in 1h 05m 24s\n- openstack-ansible-deploy-aio_metal_congress-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66e26edb18f24c8f9ac49504f47df92f : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c03e55efad89cadd86def55164343d49972a441b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-29 21:56:26.000000000","message":"Patch Set 1: Code-Review-1\n\nwhy don\u0027t we switch the nodeset to opensuse-15 instead of removing the  job alltogether?","accounts_in_message":[],"_revision_number":1},{"id":"f408e9c393b66ab4bd07362bc68f12ceedfb361b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-30 07:13:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f03bb85d6d0e5fad5c90c7865a1beb212b555516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 08:57:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/57bac3abf6fc467bafd287f7f75f7d8c : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c171a1e3f54e5689738823906d2aee : SUCCESS in 4m 35s\n- openstack-ansible-deploy-aio_metal_congress-debian-stable https://zuul.opendev.org/t/openstack/build/e6c8cab4c1fb49df9fd0c754566aa05d : SUCCESS in 1h 08m 48s\n- openstack-ansible-deploy-aio_metal_congress-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c08c0d48a76e4e498f692c87a8582e2d : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7789e4521ed870817f482d26d835e4dbc9550c90","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-30 12:35:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"835229df19d19b177b75638ae0a7f3473fda970a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-30 13:36:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"471093d7273f1a1a58d80c65b42b73adeab89c90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 13:36:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0dca936cd83c80acda5f80e9d5faa2e98edc8ddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:46:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8bd81014f57041f49bcba557566bd4fc : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbcf854bfc0f4401b44a62daf4bf7778 : SUCCESS in 2m 59s\n- openstack-ansible-deploy-aio_metal_congress-debian-stable https://zuul.opendev.org/t/openstack/build/69fe4a5faae94e878a5ef16a35a3e723 : SUCCESS in 1h 05m 34s\n- openstack-ansible-deploy-aio_metal_congress-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfaf0ba0325d45ccb2fbd3fd33c66f1b : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":2},{"id":"38e9cb056128d6d46aa4e8b7ba6dcb37ae1cb7f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:46:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5e1c5130e262c7f531d6a0ddcdb688f9f8a7060b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:50:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1423d2b1e7bf4c7aa21e85e111c93188 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"526ec716bc6095c42cb15c99833c484e4e341063","revisions":{"4ae518bafec970ef3030c4bffb8c5b1c6fdb602b":{"kind":"REWORK","_number":1,"created":"2019-10-29 20:36:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/68/691968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_congress","ref":"refs/changes/68/691968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/68/691968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/68/691968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/68/691968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/68/691968/1"}}},"commit":{"parents":[{"commit":"9ce64abdba5e4dd75c68802b6b8fde0b467477cb","subject":"Convert dynamic includes to static imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/9ce64abdba5e4dd75c68802b6b8fde0b467477cb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-29 20:33:41.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-29 20:36:31.000000000","tz":60},"subject":"Remove opensuse-150 job","message":"Remove opensuse-150 job\n\nRemove the opensuse-150 job, it is not run anymore on master and was\nremoved there (together with others) since it did not test what it\nshould test.\nWe\u0027re retiring opensuse-150 in OpenStack Infra, so remove the job\ncompletly on this branch as well.\n\nThis is a partial backport of change https://review.opendev.org/660071.\n\nChange-Id: Ia6e1341dd93a2cbf78179d1064b924b4d3126470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/4ae518bafec970ef3030c4bffb8c5b1c6fdb602b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/4ae518bafec970ef3030c4bffb8c5b1c6fdb602b"}]},"branch":"refs/heads/stable/stein"},"526ec716bc6095c42cb15c99833c484e4e341063":{"kind":"REWORK","_number":2,"created":"2019-10-30 07:13:12.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/68/691968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_congress","ref":"refs/changes/68/691968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/68/691968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/68/691968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/68/691968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_congress refs/changes/68/691968/2"}}},"commit":{"parents":[{"commit":"9ce64abdba5e4dd75c68802b6b8fde0b467477cb","subject":"Convert dynamic includes to static imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/9ce64abdba5e4dd75c68802b6b8fde0b467477cb"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-29 20:33:41.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-30 07:12:40.000000000","tz":60},"subject":"Switch to opensuse-15","message":"Switch to opensuse-15\n\nopensuse-150 nodeset is referring to openSUSE 15.0, which is still in\nmaintenance but openSUSE 15.1 has been released already. \"opensuse-15\"\nis going to refer to the \"latest openSUSE 15.x\" build released and\nworking for OpenStack going forward, so add this nodeset and use\nit by default going forward.\n\nNote this is not a backport, master has changed the job setup completly,\ninstead it just updates the jobs.\n\nChange-Id: Ia6e1341dd93a2cbf78179d1064b924b4d3126470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/526ec716bc6095c42cb15c99833c484e4e341063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_congress/commit/526ec716bc6095c42cb15c99833c484e4e341063"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
