)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~1003641","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I4eff2677aff04f756478dd56d42e1b654718072b","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"osa/pep-508","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-10-06 08:47:09.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-10-06 08:47:09.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-10-05 12:27:10.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I4eff2677aff04f756478dd56d42e1b654718072b","subject":"Use Direct URL syntax for pip packages","status":"MERGED","created":"2026-09-03 10:04:36.000000000","updated":"2026-10-06 08:48:29.000000000","submitted":"2026-10-06 08:47:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003641-osa/pep-508","meta_rev_id":"96ed1585702b0b55434c7ee1d027d526ed24adb1","_number":1003641,"virtual_id_number":1003641,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-06 08:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-10-03 05:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-10-05 12:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-10-05 12:27:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 16:04:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-02 12:11:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-10-03 05:41:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-10-03 05:41:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-10-03 05:41:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"53775b567a92b8b53de2ac8edd6d8852a35a8290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-03 10:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46d21b53afef574f070a59c839de574f2e7c6f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:04:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41bfa8829f9b42369aa278df26965b3c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4851c547c2644309b1fa6df8516e8ddd : SUCCESS in 13m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edcff68cb6d24f829e2e1c164cc4285a : SUCCESS in 1h 31m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/91b98c949910468684d79b71c461a395 : SUCCESS in 1h 03m 57s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/01fa706ebfc84c37b7cd38966270b8c7 : SUCCESS in 1h 22m 15s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/537220920128475682b30a3337558f7f : FAILURE in 1h 19m 23s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cb017a4ea68e487791d80e8968139f96 : SUCCESS in 1h 24m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4d4d6051d065471b923dce98ec3e3a2c : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4025df04f052412d82537d4d1d35f6b7 : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/491e4df7f89e48b2a0ec7c3af6ada291 : SUCCESS in 1h 17m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3aab4bfee8548ce97ce5f7c140bb32f : SUCCESS in 1h 41m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6f11bcc0f6de41f49be13d20025fd28a : SUCCESS in 2h 15m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6cc55a11582426cb6894a35dc1d728f : SUCCESS in 1h 55m 44s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41f32969873840139ed69e25e7432325 : SUCCESS in 2h 39m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309dbfab4025439481bc5b2edf2a2f10 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec6237c5cbf40e986fe13fc65f1be4e : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3de43109d2418c529dd4a765d9a819747754a3d7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-02 12:11:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"359ece5bd86bb32ff4b0ca95bb56283277b39785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 14:53:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e80d75ef93bf4a529d173c8de883a1d9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35b6400141684e33aa7149edf7ffee8f : SUCCESS in 14m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8de0e2fc392c4b55b5bcbdcdacdb5ec0 : SUCCESS in 52m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/24ab75c71c8e471ba739cad2241c099b : SUCCESS in 53m 50s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e77558fd8a5940339b69e5b874b0932e : SUCCESS in 1h 06m 06s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e15b5f5ebfa74d01aba9cecb848309b9 : SUCCESS in 1h 44m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5e14c247139e4682be69a34d94338027 : SUCCESS in 51m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e1785d1e910d425aa32c54a62dd87297 : SUCCESS in 53m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3365e9116b8a41dbab5b937bdedd3e34 : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/093a35330e354ca0b00d2b000f090cda : SUCCESS in 1h 35m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff5806d521234cb3931ef75e841c7b9e : SUCCESS in 1h 19m 30s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/178995679a6f4d5fa77215b482412b89 : SUCCESS in 2h 17m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/697962c5c43e466b80004298c4d35d73 : SUCCESS in 1h 23m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ea5b231c1094a3fb7d8dd9e06eb6161 : SUCCESS in 2h 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b448f6e21a99458f9b62e17bd7019fac : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/088629e756aa4f83a192b70b7e7163ed : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f8e7a06794eacac8d1fc5daf2c9fd145bd5a1d81","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-03 05:41:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f292fb1c485d65bf1a07aad94bc8e4806b0a2ffa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-10-05 12:27:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a23ba9e71a1450d9745eef6f51762af4886be998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 12:29:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a29575bfd69e60e93842d7917f9aa7bb5f5afdbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 14:19:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e11686e3eb874a349299f5b8708397de\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb11553fbb0141b2aaa2ed90a193e933 : SUCCESS in 10m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ed4c4fe4b854d74845e51f5bcaf34dd : SUCCESS in 1h 31m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/66a4ac0b800a4ba5b74f3041a35f779b : SUCCESS in 1h 29m 01s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3f868e54cb5341f6999767e590b4f2f6 : SUCCESS in 56m 09s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/6e06fcac77c94b52b309c5d4108121f4 : FAILURE in 1h 26m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fcb0befecb8e49c195044789a9407f4d : SUCCESS in 1h 24m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/698f400237b04df18eadb607553cedd7 : SUCCESS in 1h 42m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb496c1c3b6940ecac0d4794502e4e9a : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b7aea239e9f44998840c5affcdcc79bf : SUCCESS in 1h 35m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dad4eb501044b6a981277df10dd62da : SUCCESS in 1h 45m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81c6e53518a49b2ac50c8c8dfd628ae : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ccfafc49c44ffba2c9487168f6720a : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c5e3b2c0026d6119f6cd0e3b9981b819aa40b104","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-10-05 16:28:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db162e36900c85b4c71306e6402164b732138372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 19:32:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e839a9513728462ca96d508357533a68\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfd69d73636249d1aee8b76aaa83bfb8 : SUCCESS in 11m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7215e953784a4762bf1cf0b576dc0f19 : SUCCESS in 1h 31m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/efc0ad06e0fd4659a495fbfc200024df : SUCCESS in 1h 35m 20s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/60d24ff705f043948e04dcc58dd2d778 : SUCCESS in 1h 32m 15s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f7d684f1e7834b49bf982bac7ae0437f : FAILURE in 1h 25m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/038f99a45d6841cc91f82fa532ee8117 : SUCCESS in 54m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a20ebc50ed80404299f54c83ff13d1c1 : FAILURE in 1h 30m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ff9cc1da4b949a49a9c86b3d333d907 : SUCCESS in 1h 32m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6bd29d0aa98745d9b0be4055b63b35b2 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03a7adca441646a38aa1f48ad9173fce : SUCCESS in 1h 51m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f3f8ddbce85041c9b80e0fab88b158c1 : SUCCESS in 1h 45m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b42f963b0f34247980bd368814505ba : SUCCESS in 2h 56m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86ab3f65efb14c62abb1af843d4e15c3 : SUCCESS in 2h 40m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cfdef2511cb4914a1d11877d3d62125 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7567e3188754ed1bb22bb0fe4894339 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e5b6047df1c4b3ea28f933a28d554ec92cf84761","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-10-06 04:06:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc742599974c585a8a8749baadbb23048d33e80e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 06:59:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bf751d76602424f8433788a325968cc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f8389c326264213965f3e412c52b773 : SUCCESS in 10m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e67a1450798e4582a3abef6b4efa421b : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/79fa66ef797845fd8bcb0b3f25e14d88 : SUCCESS in 50m 37s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7b2fb61ebaeb4de9993791006ae56e3e : SUCCESS in 1h 27m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a25ad56a981d49d19dfb9e74aa7a3d2d : SUCCESS in 1h 19m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/20f55e8426d04a5b91e3c1795cb60a33 : SUCCESS in 1h 38m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d7acbee30d124fc1b60737d4f9c0f02e : SUCCESS in 52m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db5afcd33bae4500a57e6881d6535d75 : SUCCESS in 49m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d41fcd3de7d34803858e3bea2a82fcf0 : SUCCESS in 45m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b84274b9fd840fd9582c738515f131c : SUCCESS in 1h 35m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a1e7c414c1a9472a8fe58462734f4492 : SUCCESS in 2h 36m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a41292b1e65f4a8fb2101657d906ee0d : SUCCESS in 2h 32m 43s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1e28d6704604eb59bf0c7da29637a49 : SUCCESS in 2h 36m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cac3905eb344d49a7e3d8222d23c0b8 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c82c23b973da4987ad85b53f3a7aa0f9 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7fe4453d2a78c78fd8ec99f24b0a4ea46227ba7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 07:00:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8ae93f6de227d383683660bd011ddba04209088","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 08:47:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/449a3262f9b64fc18b94457fd2986a05\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec501ddd8b4649ce829f2887284e0795 : SUCCESS in 9m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7103bc0661e4b41bfa1afe689daf8da : SUCCESS in 1h 31m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e1052aaad41e4f4fac036df561173e3f : SUCCESS in 1h 25m 17s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/cb49433e17534d048f26dc27e82f0fb2 : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f4193b01b9ea433faee647f8917b4f37 : SUCCESS in 1h 22m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c3fdbdbf432c4c19b16ecf38a3a38c29 : SUCCESS in 1h 25m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/52c497123c0e45e9b0b60efb4f50a815 : SUCCESS in 1h 44m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72e80fea42fc4fe199203c3c86cfef3b : SUCCESS in 1h 37m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7cbcea01b9524b7cbf762596ba24e0e5 : SUCCESS in 1h 32m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d74aca6c7dac4f6b8c2e786ff0f17fdd : SUCCESS in 1h 33m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21460d880a8149168cfc8e38d6c7d33b : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff4f7acd8174e529ca824ca482b1c19 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fbcd00f942c0ca0e18d2523b24355a6af720e79a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 08:47:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"96ed1585702b0b55434c7ee1d027d526ed24adb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 08:48:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/911874cb9e944354be3831d3b60fbd4f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cec5bad54b6466baf9fad7de4afbe94 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5363ea0875e28be5401410585f5ec1a98256985","revisions":{"c5363ea0875e28be5401410585f5ec1a98256985":{"kind":"REWORK","_number":1,"created":"2026-09-03 10:04:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/1003641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/41/1003641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/1003641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/1003641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/1003641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/41/1003641/1"}}},"commit":{"parents":[{"commit":"a5334bd0d12ebc5822ad653fea8d9645cd09e992","subject":"Merge \"Clean-up designate config file and define workers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/a5334bd0d12ebc5822ad653fea8d9645cd09e992"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-03 10:04:28.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-03 10:04:28.000000000","tz":120},"subject":"Use Direct URL syntax for pip packages","message":"Use Direct URL syntax for pip packages\n\nUsage of #egg has been deprecated and supplying extra requirements\nfor the egg is no longer supported since pip 26.0 [1]\n\nUsage of Direct URL is conformant with PEP 508 [2]\n\n[1] https://pip.pypa.io/en/latest/news/#v26-0\n[2] https://peps.python.org/pep-0508/\n\nChange-Id: I4eff2677aff04f756478dd56d42e1b654718072b\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c5363ea0875e28be5401410585f5ec1a98256985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c5363ea0875e28be5401410585f5ec1a98256985"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
