)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~654356","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~Ideb85d8b85f47bd1cb373ddea780bc8f051b1dbd","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ideb85d8b85f47bd1cb373ddea780bc8f051b1dbd","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:14:04.000000000","updated":"2019-04-22 12:25:47.000000000","submitted":"2019-04-22 12:25:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654356-1555935947986-1d1e633c","meta_rev_id":"b26cabe39eb1f18d3cf557c54c72836971e35fa2","_number":654356,"virtual_id_number":654356,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 12:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 10:53:07.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 10:53:07.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:53:07.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 12:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2be09e0423693953aa812105ed10f3ee79efa9a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:14:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77d8deb487b8eb6549888d4043886c020543bcfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:53:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/56/654356/1/check/openstack-ansible-linters/21c1a10/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/56/654356/1/check/openstack-tox-docs/9336b12/html/ : SUCCESS in 2m 50s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/56/654356/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/2f0a64c/ : SUCCESS in 1h 20m 27s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/56/654356/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/b811e24/ : SUCCESS in 1h 03m 21s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/56/654356/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/61c0dd9/ : SUCCESS in 1h 00m 24s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/56/654356/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/1e0f978/ : SUCCESS in 1h 24m 34s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/56/654356/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/9a14d49/ : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":1},{"id":"67c8c01bffecfd7ba6439b535f6b7a4b12a0dbb0","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:53:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b477d0258de53f062c70fe8791bb2033abebe448","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:53:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4dc5662edf6ed1c0fbf4a4813547e2075614dfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:25:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f1ffef2090cef482e21449d6e7b8173fd4fea739","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:25:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/56/654356/1/gate/openstack-ansible-linters/e4964bc/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/56/654356/1/gate/openstack-tox-docs/b1b20bd/html/ : SUCCESS in 2m 18s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/56/654356/1/gate/openstack-ansible-deploy-aio_metal_designate-centos-7/ddebb39/ : SUCCESS in 1h 19m 32s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/56/654356/1/gate/openstack-ansible-deploy-aio_metal_designate-debian-stable/30e4ed9/ : SUCCESS in 58m 42s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/56/654356/1/gate/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/87f8598/ : SUCCESS in 56m 37s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/56/654356/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/0801a22/ : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/56/654356/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/f2080f3/ : SUCCESS in 1h 03m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af3c12307801b2b8e98a3163b3702c1b501e7e95","revisions":{"af3c12307801b2b8e98a3163b3702c1b501e7e95":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:14:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/654356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/56/654356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/56/654356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/56/654356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/56/654356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/56/654356/1"}}},"commit":{"parents":[{"commit":"0db848cc1a307b6d5eb2a9a7708d2eb8b2715652","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/0db848cc1a307b6d5eb2a9a7708d2eb8b2715652"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:14:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:14:02.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ideb85d8b85f47bd1cb373ddea780bc8f051b1dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/af3c12307801b2b8e98a3163b3702c1b501e7e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/af3c12307801b2b8e98a3163b3702c1b501e7e95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
