)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~654510","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Fstein~I01b77a8c548573162971f67e64451d34b2ff5f94","project":"openstack/openstack-ansible-os_designate","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I01b77a8c548573162971f67e64451d34b2ff5f94","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 15:38:05.000000000","updated":"2019-04-22 22:20:24.000000000","submitted":"2019-04-22 22:20:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654510-1555971624700-eaa21e4c","meta_rev_id":"723aaf7b5eea203aa92f5aca0a89df7b64c6b278","_number":654510,"virtual_id_number":654510,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-22 22:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-22 16:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-22 16:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:10:27.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-22 22:20:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c775afd3b3183319c3eb5ead97d3e2970fd7a052","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:38:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d19d39612e95ebdc039e3ea9dadb6efaa86efbb2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:10:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa2c78037a753464975170dccad304df6127e4b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:29:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/10/654510/1/check/openstack-ansible-linters/cac2fe3/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/10/654510/1/check/openstack-tox-docs/4421323/html/ : SUCCESS in 2m 42s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/0f65e2f/ : FAILURE in 32m 17s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/9b591db/ : FAILURE in 23m 29s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/49e7059/ : FAILURE in 21m 25s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/7b2b595/ : FAILURE in 30m 07s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/9ebbf8d/ : FAILURE in 20m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c7f97929b915fbcabece6a139bdc08f9fe803d2a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 19:11:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70ed8ea33922c0b67e61ffac12ca31907887c9ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:42:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/654510/1/check/openstack-ansible-linters/e52a7e6/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/10/654510/1/check/openstack-tox-docs/1f18e58/html/ : SUCCESS in 2m 59s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/a6b9719/ : SUCCESS in 1h 17m 53s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/d77b69b/ : SUCCESS in 1h 01m 20s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/9fb9f49/ : SUCCESS in 1h 01m 54s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/168510e/ : SUCCESS in 1h 21m 27s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/10/654510/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/45ba9a7/ : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":1},{"id":"10481036d8dd977b301f367b95f7f39b2eead2b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:43:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e932b93d556e525676c92f4da4cd85fcde6b70dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:20:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/654510/1/gate/openstack-ansible-linters/2c8dc9d/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/10/654510/1/gate/openstack-tox-docs/39d472f/html/ : SUCCESS in 2m 57s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/10/654510/1/gate/openstack-ansible-deploy-aio_metal_designate-centos-7/1098219/ : SUCCESS in 1h 30m 17s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/10/654510/1/gate/openstack-ansible-deploy-aio_metal_designate-debian-stable/64cc4c3/ : SUCCESS in 1h 02m 28s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/10/654510/1/gate/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/544dbde/ : SUCCESS in 1h 02m 28s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/10/654510/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/665cc0d/ : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/10/654510/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/9286363/ : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":1},{"id":"44005f43c70b33cc41476b0965ef244cd5c000b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:20:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94c7c3ef223519db31eadbe339734464ab73a5fe","revisions":{"94c7c3ef223519db31eadbe339734464ab73a5fe":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:38:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/654510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/10/654510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/10/654510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/10/654510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/10/654510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/10/654510/1"}}},"commit":{"parents":[{"commit":"c6bd89293fb20511bd9ff5d915a57ec3d8342786","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c6bd89293fb20511bd9ff5d915a57ec3d8342786"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:38:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:38:03.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I01b77a8c548573162971f67e64451d34b2ff5f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/94c7c3ef223519db31eadbe339734464ab73a5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/94c7c3ef223519db31eadbe339734464ab73a5fe"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
