)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~654905","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Fstein~I297c3d8e57f0c05c7c27c795683a7db91f6fb666","project":"openstack/openstack-ansible-os_designate","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I297c3d8e57f0c05c7c27c795683a7db91f6fb666","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-23 09:23:56.000000000","updated":"2019-04-24 00:29:26.000000000","submitted":"2019-04-24 00:29:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654905-1556065766692-776347d6","meta_rev_id":"35752ec0e6103db04a58b696395cab8b87149d12","_number":654905,"virtual_id_number":654905,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-24 00:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-23 20:23:25.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-23 10:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-23 10:50:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 10:50:43.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-23 20:23:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-24 00:29:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fd5d6eb676ce14f7d372ec5971c1061b6dd9181","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-23 09:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50e7f50633e43e9bbfa6bd73e2170357fba12625","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-23 10:50:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7410a330abd5a4f4c93a65c76d762796400368f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:56:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/05/654905/1/check/openstack-ansible-linters/24b5c46/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/05/654905/1/check/openstack-tox-docs/f1364ce/html/ : SUCCESS in 3m 34s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/7beeac7/ : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/960fd5b/ : SUCCESS in 1h 00m 57s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/bd40671/ : SUCCESS in 1h 00m 40s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/612219b/ : SUCCESS in 1h 22m 03s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/d543325/ : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":1},{"id":"20e1c6efd254d1e18d43ba6455c926b255ac2c86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:56:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87614546b84460b2c261d9a11763cada2a459d0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:11:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/05/654905/1/gate/openstack-ansible-linters/3172fa6/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/05/654905/1/gate/openstack-tox-docs/b844fdd/html/ : SUCCESS in 2m 17s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 finger://ze09.openstack.org/a1b535ed41454309988171fdc80c7111 : RETRY_LIMIT in 43m 26s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable finger://ze01.openstack.org/07cc76a5d8f0419b90d2553175ec0426 : RETRY_LIMIT in 44m 04s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic finger://ze12.openstack.org/30e8cc6ac4094cde9953d51493decc37 : RETRY_LIMIT in 45m 58s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 finger://ze06.openstack.org/f007ec66d973434380068f20943ff54a : RETRY_LIMIT in 41m 10s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic finger://ze09.openstack.org/d3079fd8afd04b4583e2842fe9448353 : RETRY_LIMIT in 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4d9311d44b02288aa68df47d709a3ebdc69964af","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 20:23:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5b3e61a79a8fc86e3e71e89da8b8874da21e415","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 22:50:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/05/654905/1/check/openstack-ansible-linters/93688d6/ : SUCCESS in 8m 32s\n- openstack-tox-docs http://logs.openstack.org/05/654905/1/check/openstack-tox-docs/f7f3785/html/ : SUCCESS in 3m 38s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/bffce41/ : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/08c0c4d/ : SUCCESS in 58m 54s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/89d400a/ : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/a830816/ : SUCCESS in 1h 34m 56s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/05/654905/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/d49234c/ : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0ff22107b46fbb30733fc5eb0cd9f6dc8b38f309","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 22:50:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69524a8baad91f2d750a01ad49436c0da344c3a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:29:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"652108c1f53eac86b23ab4079774463562b7b997","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:29:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/05/654905/1/gate/openstack-ansible-linters/f784994/ : SUCCESS in 5m 56s\n- openstack-tox-docs http://logs.openstack.org/05/654905/1/gate/openstack-tox-docs/fd064da/html/ : SUCCESS in 3m 11s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/05/654905/1/gate/openstack-ansible-deploy-aio_metal_designate-centos-7/ccee6a9/ : SUCCESS in 1h 25m 49s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/05/654905/1/gate/openstack-ansible-deploy-aio_metal_designate-debian-stable/49f5591/ : SUCCESS in 1h 06m 54s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/05/654905/1/gate/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/c3989fb/ : SUCCESS in 1h 00m 25s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/05/654905/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/c55801e/ : SUCCESS in 1h 18m 27s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/05/654905/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/879c8e6/ : SUCCESS in 1h 11m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7eb6a3417db3e60b0ec90c129e3daafc86b9e3bb","revisions":{"7eb6a3417db3e60b0ec90c129e3daafc86b9e3bb":{"kind":"REWORK","_number":1,"created":"2019-04-23 09:23:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/654905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/05/654905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/05/654905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/05/654905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/05/654905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/05/654905/1"}}},"commit":{"parents":[{"commit":"94c7c3ef223519db31eadbe339734464ab73a5fe","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/94c7c3ef223519db31eadbe339734464ab73a5fe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:23:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:23:55.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I297c3d8e57f0c05c7c27c795683a7db91f6fb666\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/7eb6a3417db3e60b0ec90c129e3daafc86b9e3bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/7eb6a3417db3e60b0ec90c129e3daafc86b9e3bb"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
