)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~657251","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~Ia06f067b5bdd9fc6c29ebd49c699d49a83d9634e","project":"openstack/openstack-ansible-os_designate","branch":"master","hashtags":[],"change_id":"Ia06f067b5bdd9fc6c29ebd49c699d49a83d9634e","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-05-05 12:36:24.000000000","updated":"2019-07-16 17:43:10.000000000","submitted":"2019-07-16 17:43:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657251-1563298990070-5870c8eb","meta_rev_id":"feaac942bfcfb9e9b4fc51705daf38e5725f2fb9","_number":657251,"virtual_id_number":657251,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-16 17:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":1,"date":"2019-07-13 22:01:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":1,"date":"2019-07-14 12:32:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-07-16 14:23:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-14 13:43:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":2,"date":"2019-07-16 15:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":1,"date":"2019-07-16 15:58:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-13 22:01:12.000000000","updated_by":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"reviewer":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"state":"REVIEWER"},{"updated":"2019-07-14 12:32:24.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-14 13:43:50.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-16 14:23:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-16 15:58:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-16 17:43:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3740ccb8021870bc859624e86cca6848e920745b","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 12:36:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddd5da9dc0275096b3ddc9803c9031568aa48d07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 14:08:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/657251/1/check/openstack-ansible-linters/3298688/ : FAILURE in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/51/657251/1/check/openstack-tox-docs/1953640/html/ : SUCCESS in 2m 06s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/51/657251/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/66b1d5b/ : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/51/657251/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/c13a7f1/ : SUCCESS in 58m 23s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/51/657251/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/2b54966/ : SUCCESS in 1h 07m 15s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/51/657251/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/26ef826/ : SUCCESS in 1h 19m 06s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/51/657251/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/a9746c8/ : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9743da22d1285368ea20f0586bca3be4dc5d165f","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 11:08:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ccc8b13114ad427882664f5b5eda8503e7b9069","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:32:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/657251/2/check/openstack-ansible-linters/21f62d2/ : FAILURE in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/51/657251/2/check/openstack-tox-docs/23c6b38/html/ : SUCCESS in 3m 30s\n- build-openstack-releasenotes http://logs.openstack.org/51/657251/2/check/build-openstack-releasenotes/cde63b1/html/ : SUCCESS in 3m 16s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/51/657251/2/check/openstack-ansible-deploy-aio_metal_designate-centos-7/75481eb/ : SUCCESS in 1h 18m 57s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/51/657251/2/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/122120c/ : SUCCESS in 1h 05m 33s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/51/657251/2/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/bef4c6b/ : SUCCESS in 56m 50s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/51/657251/2/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/bead6bc/ : SUCCESS in 1h 16m 29s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/51/657251/2/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/4390905/ : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":2},{"id":"7b41952497a7f5616989954b81143c4b53f19339","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-10 14:06:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ede5ea842dc1f74a15ce6fc1192097323574071","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:44:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/51/657251/3/check/openstack-ansible-linters/a25ff0a/ : FAILURE in 5m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/51/657251/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/fd85610/ : SUCCESS in 1h 17m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/51/657251/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/202e318/ : FAILURE in 30m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/51/657251/3/check/openstack-ansible-deploy-aio_metal-debian-stable/2fc5cbe/ : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/51/657251/3/check/openstack-ansible-deploy-aio_metal-centos-7/8c6dbdf/ : SUCCESS in 1h 47m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/51/657251/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/52ff669/ : SUCCESS in 2h 25m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/51/657251/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/305b997/ : SUCCESS in 1h 03m 15s\n- openstack-tox-docs http://logs.openstack.org/51/657251/3/check/openstack-tox-docs/c1f2fe3/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/51/657251/3/check/build-openstack-releasenotes/b7ed320/html/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c9c17f54e7945c378b3887e4facc77e6ec1fe304","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-13 19:31:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a1cb8c074370dfac4d67a97742730f1c9389649a","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-13 19:55:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ab2000933384db2aff57210c6c91ea711a7a9ed7","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-13 20:29:07.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"258639e2893dc9f1e1a541fc4ea06bf693f9694b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 22:00:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/51/657251/5/check/openstack-ansible-linters/19cc525/ : SUCCESS in 4m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/51/657251/5/check/openstack-ansible-deploy-aio_distro_metal-centos-7/353a301/ : SUCCESS in 1h 06m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/51/657251/5/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/708d525/ : SUCCESS in 56m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/51/657251/5/check/openstack-ansible-deploy-aio_metal-debian-stable/74fc8ca/ : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/51/657251/5/check/openstack-ansible-deploy-aio_metal-centos-7/4c645fa/ : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/51/657251/5/check/openstack-ansible-deploy-aio_metal-opensuse-15/900b1d8/ : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/51/657251/5/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d405288/ : SUCCESS in 1h 01m 55s\n- openstack-tox-docs http://logs.openstack.org/51/657251/5/check/openstack-tox-docs/fe8a81f/html/ : SUCCESS in 2m 22s\n- build-openstack-releasenotes http://logs.openstack.org/51/657251/5/check/build-openstack-releasenotes/d5b024a/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":5},{"id":"3f09b8dae10066147c31a429116117895055c2fa","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-13 22:01:12.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1be60e1f9baa46f661a728a771b7d292e92d1842","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 12:32:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"893219217f125e39522b70d6727e6a0546b6220e","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-14 13:43:50.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"78eb4930cf0217060a41945b08d2cebe3c27fad1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 14:23:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f276712470d86ae8bbc64955f5f3c37a8836172a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-16 15:58:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"db82404c45835312b417d184fc91704a6fdb268c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:59:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a5f680b1f5cbdb02866d34a8f077d4016144e2d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:43:09.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/51/657251/5/gate/openstack-ansible-linters/fa719dd/ : SUCCESS in 8m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/51/657251/5/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/f77c1fa/ : SUCCESS in 1h 06m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/51/657251/5/gate/openstack-ansible-deploy-aio_metal-debian-stable/6e0594e/ : SUCCESS in 1h 08m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/51/657251/5/gate/openstack-ansible-deploy-aio_metal-centos-7/5ff42a1/ : SUCCESS in 1h 22m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/51/657251/5/gate/openstack-ansible-deploy-aio_metal-opensuse-15/db5c9e9/ : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/51/657251/5/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8ba0aa1/ : SUCCESS in 1h 13m 09s\n- openstack-tox-docs http://logs.openstack.org/51/657251/5/gate/openstack-tox-docs/b0740a8/html/ : SUCCESS in 3m 04s\n- build-openstack-releasenotes http://logs.openstack.org/51/657251/5/gate/build-openstack-releasenotes/eb8de48/html/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":5},{"id":"feaac942bfcfb9e9b4fc51705daf38e5725f2fb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:43:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"066c4025f5c8e0e3c755ade572b128c60c7f4ce0","revisions":{"10c03f2d301f1e1a218995c8c56d0e9c47d7fbde":{"kind":"REWORK","_number":1,"created":"2019-05-05 12:36:24.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/51/657251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/51/657251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/1"}}},"commit":{"parents":[{"commit":"af3c12307801b2b8e98a3163b3702c1b501e7e95","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/af3c12307801b2b8e98a3163b3702c1b501e7e95"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:36:04.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:36:21.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia06f067b5bdd9fc6c29ebd49c699d49a83d9634e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/10c03f2d301f1e1a218995c8c56d0e9c47d7fbde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/10c03f2d301f1e1a218995c8c56d0e9c47d7fbde"}]},"branch":"refs/heads/master"},"2cea67c433bc73fc74213ffc14643e0fb80e998f":{"kind":"REWORK","_number":2,"created":"2019-05-06 11:08:33.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/51/657251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/51/657251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/2"}}},"commit":{"parents":[{"commit":"af3c12307801b2b8e98a3163b3702c1b501e7e95","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/af3c12307801b2b8e98a3163b3702c1b501e7e95"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:36:04.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 11:08:20.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia06f067b5bdd9fc6c29ebd49c699d49a83d9634e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/2cea67c433bc73fc74213ffc14643e0fb80e998f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/2cea67c433bc73fc74213ffc14643e0fb80e998f"}]},"branch":"refs/heads/master"},"43223a657cc9ffaa88e5e64c12f3fe8874904143":{"kind":"REWORK","_number":3,"created":"2019-07-10 14:06:45.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/51/657251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/51/657251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/3"}}},"commit":{"parents":[{"commit":"af3c12307801b2b8e98a3163b3702c1b501e7e95","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/af3c12307801b2b8e98a3163b3702c1b501e7e95"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:36:04.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 14:06:18.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia06f067b5bdd9fc6c29ebd49c699d49a83d9634e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/43223a657cc9ffaa88e5e64c12f3fe8874904143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/43223a657cc9ffaa88e5e64c12f3fe8874904143"}]},"branch":"refs/heads/master"},"13f80e687565b8c8c0290b5a49db6bf14d7960f9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-13 19:55:03.000000000","uploader":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"ref":"refs/changes/51/657251/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/51/657251/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/4"}}},"commit":{"parents":[{"commit":"092374b4d0f6d83427fa370fceb81ddd09306119","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/092374b4d0f6d83427fa370fceb81ddd09306119"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:36:04.000000000","tz":480},"committer":{"name":"Jonathan Herlin","email":"jonte@jherlin.se","date":"2019-07-13 19:55:03.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia06f067b5bdd9fc6c29ebd49c699d49a83d9634e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/13f80e687565b8c8c0290b5a49db6bf14d7960f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/13f80e687565b8c8c0290b5a49db6bf14d7960f9"}]},"branch":"refs/heads/master"},"066c4025f5c8e0e3c755ade572b128c60c7f4ce0":{"kind":"REWORK","_number":5,"created":"2019-07-13 20:29:07.000000000","uploader":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"ref":"refs/changes/51/657251/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/51/657251/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/51/657251/5"}}},"commit":{"parents":[{"commit":"092374b4d0f6d83427fa370fceb81ddd09306119","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/092374b4d0f6d83427fa370fceb81ddd09306119"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 12:36:04.000000000","tz":480},"committer":{"name":"Jonathan Herlin","email":"jonte@jherlin.se","date":"2019-07-13 20:28:24.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia06f067b5bdd9fc6c29ebd49c699d49a83d9634e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/066c4025f5c8e0e3c755ade572b128c60c7f4ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/066c4025f5c8e0e3c755ade572b128c60c7f4ce0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
