)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~663564","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I023947ba2c2ff8a59cdd98c19bd2770470b1751b","project":"openstack/openstack-ansible-os_designate","branch":"master","hashtags":[],"change_id":"I023947ba2c2ff8a59cdd98c19bd2770470b1751b","subject":"Cap sphinx for py2 to match global requirements","status":"MERGED","created":"2019-06-06 09:36:21.000000000","updated":"2019-07-10 18:54:38.000000000","submitted":"2019-07-10 18:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663564-1562784878306-0696cc2b","meta_rev_id":"7db58c50c657f4db05d3b82a0d5809b2ab6f0e7e","_number":663564,"virtual_id_number":663564,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-10 18:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-10 18:38:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-10 18:38:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 18:38:39.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 18:54:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72d4543c260a82f526ff69039d430932117b2604","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-06-06 09:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f873ef04afd87290814e8ada23a9e51acfb121d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:54:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/64/663564/1/check/requirements-check/6af8494/ : SUCCESS in 2m 11s\n- openstack-ansible-linters http://logs.openstack.org/64/663564/1/check/openstack-ansible-linters/20bf7a7/ : SUCCESS in 7m 51s\n- openstack-tox-docs http://logs.openstack.org/64/663564/1/check/openstack-tox-docs/e44d3bd/html/ : SUCCESS in 2m 43s\n- build-openstack-releasenotes http://logs.openstack.org/64/663564/1/check/build-openstack-releasenotes/949dbb3/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"92993a839881934f359e7a3bc4d65dd1296b0ff4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 18:38:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8872168c66c6ab31a629aa67346159434633184","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:38:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"434ddbb1283081b186705f8818e277ae48a0592e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:54:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/64/663564/1/gate/requirements-check/bc0980a/ : SUCCESS in 2m 15s\n- openstack-ansible-linters http://logs.openstack.org/64/663564/1/gate/openstack-ansible-linters/afeaace/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/64/663564/1/gate/openstack-tox-docs/bfbe686/html/ : SUCCESS in 2m 42s\n- build-openstack-releasenotes http://logs.openstack.org/64/663564/1/gate/build-openstack-releasenotes/f9a8597/html/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7db58c50c657f4db05d3b82a0d5809b2ab6f0e7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:54:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea49d93ab2d6441b776102d2a5a29d2c496cfb52","revisions":{"ea49d93ab2d6441b776102d2a5a29d2c496cfb52":{"kind":"REWORK","_number":1,"created":"2019-06-06 09:36:21.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/64/663564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/64/663564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/64/663564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/64/663564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/64/663564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/64/663564/1"}}},"commit":{"parents":[{"commit":"7e5a171d2e5f87067e6d50d84eebc7033a74d394","subject":"db_setup: refactor database setup to a common file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/7e5a171d2e5f87067e6d50d84eebc7033a74d394"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 09:36:00.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-06-06 09:36:16.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I023947ba2c2ff8a59cdd98c19bd2770470b1751b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/ea49d93ab2d6441b776102d2a5a29d2c496cfb52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/ea49d93ab2d6441b776102d2a5a29d2c496cfb52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
