)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~663792","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I40a596803df656f5e6f75996660fe9ace685c2ed","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"ubuntu-distro-install","hashtags":[],"change_id":"I40a596803df656f5e6f75996660fe9ace685c2ed","subject":"Fix distro installs on Ubuntu","status":"MERGED","created":"2019-06-06 22:41:04.000000000","updated":"2019-06-11 21:08:16.000000000","submitted":"2019-06-10 22:57:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663792-1560207469938-4a3a791a","meta_rev_id":"fc4f4407a913dff45b62d1d894f78fc1ba1ee0e8","_number":663792,"virtual_id_number":663792,"owner":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-07 19:26:43.000000000","_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-10 22:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":2,"date":"2019-06-07 20:00:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-10 21:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-10 21:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 20:00:55.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-10 21:07:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-10 22:57:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a3f6d5539227771650499153757730d2c09051e","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-06 22:41:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86bd7db59756de799deecb4dde3ede6d2b6ab32f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:32:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/92/663792/1/check/openstack-ansible-linters/a0ad2f6/ : SUCCESS in 6m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/7958dfe/ : SUCCESS in 1h 03m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-debian-stable/0c84af0/ : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-centos-7/72d7e02/ : FAILURE in 35m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/0497f25/ : SUCCESS in 1h 15m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/36e98ec/ : FAILURE in 35m 24s\n- openstack-tox-docs http://logs.openstack.org/92/663792/1/check/openstack-tox-docs/2449be5/html/ : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"37f5b31f9bdb4afa75c069949ecb28235da25a4c","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-07 04:50:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ce83ca89b45f246fa8fc98fb6fde3870d1fdb86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 06:37:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/92/663792/1/check/openstack-ansible-linters/0c2b934/ : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/673ba08/ : SUCCESS in 59m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-debian-stable/7c87adf/ : SUCCESS in 1h 03m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-centos-7/0de99ec/ : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/a931f95/ : SUCCESS in 1h 19m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/780c671/ : FAILURE in 27m 54s\n- openstack-tox-docs http://logs.openstack.org/92/663792/1/check/openstack-tox-docs/759b546/html/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"23778a8fca23aa0685b3590aba5294a04157a96a","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-07 19:26:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e5c9831be236c13eb3dfe9d5b1d4d02c17f7b7ea","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-07 20:00:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e9be3caef8cb4004756abc683475615a4f2cab1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 22:52:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/92/663792/1/check/openstack-ansible-linters/8efe13c/ : SUCCESS in 4m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/d9bc6f1/ : SUCCESS in 1h 08m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-debian-stable/f2918ec/ : SUCCESS in 1h 33m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-centos-7/a3e9337/ : SUCCESS in 1h 22m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/6cf9511/ : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/92/663792/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/290eab1/ : SUCCESS in 1h 05m 12s\n- openstack-tox-docs http://logs.openstack.org/92/663792/1/check/openstack-tox-docs/02d9640/html/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e6425494cd3d642779c444b3d5fcf27c604bfd15","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-10 21:07:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"716fb644de7aad93858859494f1b931e389b2856","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 21:07:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eda2aaa1254ffea031fccdc3ce8e099911ea089d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 22:57:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/92/663792/1/gate/openstack-ansible-linters/1e4d4b9/ : SUCCESS in 5m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/92/663792/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/fdaec2f/ : SUCCESS in 1h 00m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/92/663792/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/9461c07/ : SUCCESS in 1h 05m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/92/663792/1/gate/openstack-ansible-deploy-aio_metal-centos-7/1f6fdee/ : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/92/663792/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/d7f02d6/ : SUCCESS in 1h 24m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/92/663792/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d7089e1/ : SUCCESS in 1h 02m 32s\n- openstack-tox-docs http://logs.openstack.org/92/663792/1/gate/openstack-tox-docs/b0a9419/html/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"99f788ec044f7c2895c6ab84a055e1b12eda6918","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 22:57:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"73ca8a45af22ddfb89d23d193888c6444cff247e","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-11 21:08:16.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f4dd7f637068dd583e5d1375ad5fb60e62e7f69c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4de0376510d26a00d06f4984c763ace92cf3b58","revisions":{"a4de0376510d26a00d06f4984c763ace92cf3b58":{"kind":"REWORK","_number":1,"created":"2019-06-06 22:41:04.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/92/663792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/92/663792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/92/663792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/92/663792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/92/663792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/92/663792/1"}}},"commit":{"parents":[{"commit":"7e5a171d2e5f87067e6d50d84eebc7033a74d394","subject":"db_setup: refactor database setup to a common file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/7e5a171d2e5f87067e6d50d84eebc7033a74d394"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-06 22:35:33.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-06 22:38:35.000000000","tz":-420},"subject":"Fix distro installs on Ubuntu","message":"Fix distro installs on Ubuntu\n\nInstall the python3-designate package and rely on the OSA provided\nservice configs instead of installing each of the individual designate\nservice packages. The \u0027Mask any conflicting units\u0027 task can also be\nremoved since the installation of unwanted services will no longer be an\nissue.\n\nChange-Id: I40a596803df656f5e6f75996660fe9ace685c2ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/a4de0376510d26a00d06f4984c763ace92cf3b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/a4de0376510d26a00d06f4984c763ace92cf3b58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
