)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~664087","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Fqueens~Idfcc1b66c379ea29e01a090c23841081d806b333","project":"openstack/openstack-ansible-os_designate","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Idfcc1b66c379ea29e01a090c23841081d806b333","subject":"Updated from OpenStack Ansible Tests","status":"ABANDONED","created":"2019-06-07 23:24:01.000000000","updated":"2020-10-13 20:36:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d103f81b37475442fcc57cc07bc04436b5acb320","_number":664087,"virtual_id_number":664087,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-08 00:29:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-11 20:54:00.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 00:29:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-11 20:54:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"2cafb9135fa32ea3f9d3df721aadb25bfcaa0a35","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:24:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a63d2d8f1147fb0b454742ad5cb0eea0956dafdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:29:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/87/664087/1/check/build-openstack-sphinx-docs/5ca7878/html/ : SUCCESS in 3m 37s\n- openstack-ansible-linters http://logs.openstack.org/87/664087/1/check/openstack-ansible-linters/ab9ad4c/ : SUCCESS in 5m 43s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/87/664087/1/check/openstack-ansible-functional-centos-7/c5e0a24/ : FAILURE in 49m 36s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/87/664087/1/check/openstack-ansible-functional-opensuse-423/7be0c74/ : FAILURE in 9m 24s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/87/664087/1/check/openstack-ansible-functional-ubuntu-xenial/b87e3b7/ : FAILURE in 30m 07s","accounts_in_message":[],"_revision_number":1},{"id":"23ca391c691bd64a92147ee76f247d25b84fdcc3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 20:54:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e4e231588272012c5296627e3bef192562e2afa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:57:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/87/664087/1/check/build-openstack-sphinx-docs/827b226/html/ : SUCCESS in 3m 24s\n- openstack-ansible-linters http://logs.openstack.org/87/664087/1/check/openstack-ansible-linters/239e6fa/ : SUCCESS in 5m 09s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/87/664087/1/check/openstack-ansible-functional-centos-7/ca49185/ : FAILURE in 51m 05s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/87/664087/1/check/openstack-ansible-functional-ubuntu-xenial/65b7ab9/ : FAILURE in 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"adfbec883d9eeb579872cc5e449738f2f9869907","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-13 20:36:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b629553bd297c844233970d3ddea2092603bc512","revisions":{"b629553bd297c844233970d3ddea2092603bc512":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:24:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/664087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/87/664087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/87/664087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/87/664087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/87/664087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/87/664087/1"}}},"commit":{"parents":[{"commit":"2d166e07585987825773abe8aee977943f04e96d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/2d166e07585987825773abe8aee977943f04e96d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:23:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:23:57.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Idfcc1b66c379ea29e01a090c23841081d806b333\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/b629553bd297c844233970d3ddea2092603bc512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/b629553bd297c844233970d3ddea2092603bc512"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
