)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~664314","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Frocky~I757db189f483f9070ea79ec3a9adf9d83d9b7d39","project":"openstack/openstack-ansible-os_designate","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I757db189f483f9070ea79ec3a9adf9d83d9b7d39","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:44:11.000000000","updated":"2019-06-10 17:59:06.000000000","submitted":"2019-06-10 17:59:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664314-1560189546735-48a02ede","meta_rev_id":"2c9ca26e0f570797c1bf265d64212664e9ca6a09","_number":664314,"virtual_id_number":664314,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 17:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:05:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:05:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:05:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 17:59:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"768e258cc1e2ba017472d8a7262f0be81e560125","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:44:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"446f4ac2956ea0b1784c2cce10ea11c1d7314bf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:05:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/14/664314/1/check/openstack-ansible-linters/d46c51b/ : SUCCESS in 4m 49s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/664314/1/check/openstack-ansible-functional-centos-7/c041ad1/ : SUCCESS in 36m 12s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/664314/1/check/openstack-ansible-functional-opensuse-423/60dee93/ : FAILURE in 7m 02s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/14/664314/1/check/openstack-ansible-functional-opensuse-150/1f0c445/ : FAILURE in 10m 29s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/664314/1/check/openstack-ansible-functional-ubuntu-bionic/208a474/ : SUCCESS in 28m 46s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/664314/1/check/openstack-ansible-functional-ubuntu-xenial/2ff3729/ : SUCCESS in 30m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/664314/1/check/build-openstack-sphinx-docs/8353896/html/ : SUCCESS in 4m 41s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/14/664314/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/75c092a/ : SUCCESS in 47m 03s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/14/664314/1/check/openstack-ansible-functional-distro_install-centos-7/c621627/ : SUCCESS in 35m 30s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/664314/1/check/openstack-ansible-functional-distro_install-opensuse-423/532ccf4/ : FAILURE in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0068f6349d1f78ba6894db6ac7d9629731cad786","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:05:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f5c23689779613e0419f821b1fe5987dacad3af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:06:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85ce246d42c06e4a18a13dfc0d22bc9715c41e85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:59:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0261cf393e83cd77846d9700e8bd2b9523058f00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:59:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/14/664314/1/gate/openstack-ansible-linters/c5a564b/ : SUCCESS in 4m 56s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/14/664314/1/gate/openstack-ansible-functional-centos-7/72fbf9d/ : SUCCESS in 39m 50s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/14/664314/1/gate/openstack-ansible-functional-opensuse-423/e07d648/ : FAILURE in 7m 31s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/14/664314/1/gate/openstack-ansible-functional-ubuntu-bionic/6d131b9/ : SUCCESS in 30m 39s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/14/664314/1/gate/openstack-ansible-functional-ubuntu-xenial/3be41e6/ : SUCCESS in 32m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/14/664314/1/gate/build-openstack-sphinx-docs/f436e1b/html/ : SUCCESS in 3m 44s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/14/664314/1/gate/openstack-ansible-functional-distro_install-ubuntu-xenial/466b478/ : SUCCESS in 42m 00s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/14/664314/1/gate/openstack-ansible-functional-distro_install-centos-7/7046bc1/ : SUCCESS in 34m 40s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/14/664314/1/gate/openstack-ansible-functional-distro_install-opensuse-423/687cae1/ : FAILURE in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d883ecfcc4da663772187d025d12d4b0b28966c","revisions":{"9d883ecfcc4da663772187d025d12d4b0b28966c":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:44:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/664314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/14/664314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/14/664314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/14/664314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/14/664314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/14/664314/1"}}},"commit":{"parents":[{"commit":"bed697bd45b031f272679bb319359d0e83e7fba6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/bed697bd45b031f272679bb319359d0e83e7fba6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:10.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I757db189f483f9070ea79ec3a9adf9d83d9b7d39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/9d883ecfcc4da663772187d025d12d4b0b28966c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/9d883ecfcc4da663772187d025d12d4b0b28966c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
