)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~666106","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I601bf4a68c1491253dfc90f36dca247ab092fa39","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I601bf4a68c1491253dfc90f36dca247ab092fa39","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-18 18:15:40.000000000","updated":"2019-06-18 21:32:32.000000000","submitted":"2019-06-18 21:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666106-1560893552351-b3e15709","meta_rev_id":"fc6260b4608511c99948028af1364e3f8e9af71c","_number":666106,"virtual_id_number":666106,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-18 21:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-18 18:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-18 19:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-18 19:50:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 18:54:35.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-18 19:50:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-18 21:32:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc8fd686e0cfc3affb885a81eba1c93d996f3c22","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 18:15:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f1d2969054ecdeb84962cef5fcbadf195631027","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-18 18:54:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b217f3f361aecd7bccfcf5f600d6b657cdcae79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:45:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/06/666106/1/check/openstack-ansible-linters/60431b0/ : SUCCESS in 5m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/666106/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/a42bb89/ : SUCCESS in 1h 12m 17s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/666106/1/check/openstack-ansible-deploy-aio_metal-debian-stable/df45021/ : SUCCESS in 1h 08m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/666106/1/check/openstack-ansible-deploy-aio_metal-centos-7/e915c25/ : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/666106/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/2166592/ : SUCCESS in 1h 11m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/666106/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8519a8f/ : SUCCESS in 1h 01m 58s\n- openstack-tox-docs http://logs.openstack.org/06/666106/1/check/openstack-tox-docs/c9c3a49/html/ : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5677c56820864ab1e04125f7d0239111dc98eca9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-18 19:50:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15cc930f3c6f0442cff728226654ded8549a4821","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:51:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57d27d8032a509bf1377fdc3510bc2ad8dcdfdc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:32:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9a9d6e441d111fec9a58728c37d42ec5e388f59b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:32:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/06/666106/1/gate/openstack-ansible-linters/0c1f17f/ : SUCCESS in 5m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/666106/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/0d0e684/ : SUCCESS in 1h 00m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/666106/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/58114b1/ : SUCCESS in 1h 09m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/666106/1/gate/openstack-ansible-deploy-aio_metal-centos-7/e18561a/ : SUCCESS in 1h 26m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/666106/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/7a2a09a/ : SUCCESS in 1h 12m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/666106/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6b79a29/ : SUCCESS in 1h 14m 52s\n- openstack-tox-docs http://logs.openstack.org/06/666106/1/gate/openstack-tox-docs/f0d71ee/html/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edbf74abe6e706684cc2e6c371612580a5ad0126","revisions":{"edbf74abe6e706684cc2e6c371612580a5ad0126":{"kind":"REWORK","_number":1,"created":"2019-06-18 18:15:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/666106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/06/666106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/06/666106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/06/666106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/06/666106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/06/666106/1"}}},"commit":{"parents":[{"commit":"e161e28b8486430186202ede49406936c9a64910","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/e161e28b8486430186202ede49406936c9a64910"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:15:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:15:39.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I601bf4a68c1491253dfc90f36dca247ab092fa39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/edbf74abe6e706684cc2e6c371612580a5ad0126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/edbf74abe6e706684cc2e6c371612580a5ad0126"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
