)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~666704","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Fstein~Ia219b95cf39f4de526c70fb53bc4bc34f9c1e85e","project":"openstack/openstack-ansible-os_designate","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia219b95cf39f4de526c70fb53bc4bc34f9c1e85e","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-20 20:04:36.000000000","updated":"2019-06-21 10:10:10.000000000","submitted":"2019-06-21 10:10:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666704-1561111810918-f027147f","meta_rev_id":"71872f2c874a134dfbea225aad7953c9a96fd5aa","_number":666704,"virtual_id_number":666704,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-21 10:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-20 20:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-21 07:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-21 07:12:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 20:08:46.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-21 07:12:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-21 10:10:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9e532effe25ba315ece5de88bfefec13b5195e0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54cb6f0f20f2647acf32032d5073007f400b44b3","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-20 20:08:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f60fd000efba0ed037b3f02950ee97c5b3d46169","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:53:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/04/666704/1/check/openstack-ansible-linters/d603ec4/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/04/666704/1/check/openstack-tox-docs/9abc634/html/ : SUCCESS in 3m 12s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/a6f6daa/ : SUCCESS in 1h 17m 12s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/f837c80/ : FAILURE in 6m 06s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/df5c645/ : SUCCESS in 58m 44s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/992ee5b/ : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/be6d536/ : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1d0db55589144a427ded374a7c6cc6fa0475713c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 22:11:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d7f3f175a0e845da7bd3dd33007104e0b58e253","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 23:45:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/04/666704/1/check/openstack-ansible-linters/cdb3bea/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/04/666704/1/check/openstack-tox-docs/e68bfc3/html/ : SUCCESS in 2m 54s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/9017aa1/ : SUCCESS in 1h 26m 34s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/f0a1916/ : SUCCESS in 1h 02m 46s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/ed19185/ : SUCCESS in 57m 42s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/799c8c1/ : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/04/666704/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/c5707d7/ : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":1},{"id":"27566f80f72d58c107b8438bd50f63231eb19ba7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-21 07:12:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e305ec24f9bc921811f89d778dee82ca4c678c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 07:12:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"031adf7ce55c2142bbd71da2b1cc03e8b905e0ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 10:10:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"03daf2a692d85d1f418dfd962b4ed76af7f29a7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 10:10:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/04/666704/1/gate/openstack-ansible-linters/94829d6/ : SUCCESS in 14m 17s\n- openstack-tox-docs http://logs.openstack.org/04/666704/1/gate/openstack-tox-docs/b23fee2/html/ : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 http://logs.openstack.org/04/666704/1/gate/openstack-ansible-deploy-aio_metal_designate-centos-7/1f20ac1/ : SUCCESS in 1h 28m 07s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable http://logs.openstack.org/04/666704/1/gate/openstack-ansible-deploy-aio_metal_designate-debian-stable/0f14f3d/ : SUCCESS in 2h 52m 11s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic http://logs.openstack.org/04/666704/1/gate/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/24d4755/ : SUCCESS in 1h 46m 25s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 http://logs.openstack.org/04/666704/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/23f63d4/ : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic http://logs.openstack.org/04/666704/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/119e84d/ : SUCCESS in 1h 47m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39638c4395d62d0e2cfb75b56606d6935648153c","revisions":{"39638c4395d62d0e2cfb75b56606d6935648153c":{"kind":"REWORK","_number":1,"created":"2019-06-20 20:04:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/666704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/04/666704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/04/666704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/04/666704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/04/666704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/04/666704/1"}}},"commit":{"parents":[{"commit":"8459828ff5ec675f171917c67f19fb1dd660a44e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/8459828ff5ec675f171917c67f19fb1dd660a44e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:04:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:04:34.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia219b95cf39f4de526c70fb53bc4bc34f9c1e85e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/39638c4395d62d0e2cfb75b56606d6935648153c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/39638c4395d62d0e2cfb75b56606d6935648153c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
