)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~673484","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~Id9e94865467a9076ee15a9caebcc2cc97847d079","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"osa/service-refactor","hashtags":[],"change_id":"Id9e94865467a9076ee15a9caebcc2cc97847d079","subject":"service_setup: refactor service setup to a single file","status":"MERGED","created":"2019-07-30 11:35:12.000000000","updated":"2019-08-02 20:55:01.000000000","submitted":"2019-08-02 20:55:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673484-1564779301749-04ad8034","meta_rev_id":"698a97fc12323b93bd9c18d11cce018fe705b00c","_number":673484,"virtual_id_number":673484,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-01 19:36:02.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-02 20:55:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-02 12:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-02 19:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-08-02 19:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 12:55:33.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-02 19:09:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-02 20:55:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cd33a097dc374082123bf71065240790ae1a45f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 11:35:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f63e422c51893aa239f0c004e49de64fd1e4b2d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 13:07:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/673484/1/check/openstack-ansible-linters/902dfd6/ : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/84/673484/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/cd66ea2/ : SUCCESS in 56m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/84/673484/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/5d29726/ : FAILURE in 51m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/84/673484/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/9e81010/ : SUCCESS in 1h 04m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/84/673484/1/check/openstack-ansible-deploy-aio_metal-debian-stable/0618d66/ : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/84/673484/1/check/openstack-ansible-deploy-aio_metal-centos-7/bf9daae/ : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/84/673484/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/32911ff/ : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/84/673484/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0973ba3/ : SUCCESS in 1h 11m 15s\n- openstack-tox-docs http://logs.openstack.org/84/673484/1/check/openstack-tox-docs/1a4e4bd/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cdca8a21b46847c0de581c1c11345cf7d47ef157","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 15:09:26.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c2d779539a4c070a070fae2b988bbf44378bca0d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-01 14:29:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac52a333b3c2e80d95369ad7600b775edcefd31a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 16:59:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/84/673484/2/check/openstack-ansible-linters/fe5775f/ : SUCCESS in 6m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/bc5f2c6/ : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/47bb8dc/ : SUCCESS in 54m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/395024c/ : SUCCESS in 49m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_metal-debian-stable/b705a9e/ : FAILURE in 1h 03m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_metal-centos-7/363ba46/ : SUCCESS in 1h 23m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/783eca6/ : SUCCESS in 1h 26m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/770579d/ : SUCCESS in 1h 01m 12s\n- openstack-tox-docs https://logs.opendev.org/84/673484/2/check/openstack-tox-docs/ec2832b/html/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"837e16a22af4f7cd0c01acf4af2d06841ab67529","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-01 19:36:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5e6f405a5df4a68b9e46f84986f91c2510e5e93b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 21:13:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/84/673484/2/check/openstack-ansible-linters/29c09d8/ : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/76b6838/ : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/d86fba0/ : SUCCESS in 58m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/c7e0806/ : SUCCESS in 46m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_metal-debian-stable/7b03f85/ : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_metal-centos-7/5b19f4f/ : SUCCESS in 1h 25m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/dbb7b7b/ : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/84/673484/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c7a34c6/ : SUCCESS in 1h 09m 30s\n- openstack-tox-docs https://logs.opendev.org/84/673484/2/check/openstack-tox-docs/d5aeb5d/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"894534649289c451311c1ab5caa8c2923c517665","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-02 12:55:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ec58da2730846f4b345c6d27872d33ba20ea6fc9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-02 19:09:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a7dfdec23afe5839882ad6e7d4134719c556094","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 19:10:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16df575ea189dc6068b6f81fd07736b773f3b56e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:55:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"295b6f36c3eda4d9d19f75dc96a22e395ec3ad42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:55:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/84/673484/2/gate/openstack-ansible-linters/f675e4f/ : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/84/673484/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/69198c1/ : SUCCESS in 1h 01m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/84/673484/2/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/d744def/ : SUCCESS in 46m 43s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/84/673484/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/12f0455/ : SUCCESS in 1h 11m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/84/673484/2/gate/openstack-ansible-deploy-aio_metal-centos-7/ee2eb89/ : SUCCESS in 1h 39m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/84/673484/2/gate/openstack-ansible-deploy-aio_metal-opensuse-15/8e5c808/ : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/84/673484/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/35ce08b/ : SUCCESS in 1h 04m 29s\n- openstack-tox-docs https://logs.opendev.org/84/673484/2/gate/openstack-tox-docs/15f2a4a/html/ : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ac2ef7a6ac1e19f55773ee6468372bcb5142756","revisions":{"380d7abacf6e085d82bcf466210906ce8a54fef2":{"kind":"REWORK","_number":1,"created":"2019-07-30 11:35:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/673484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/84/673484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/84/673484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/84/673484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/84/673484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/84/673484/1"}}},"commit":{"parents":[{"commit":"865a370027ced8240118c61dd387220b22306b03","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/865a370027ced8240118c61dd387220b22306b03"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-30 11:34:49.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-30 11:35:09.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\nservice_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Id9e94865467a9076ee15a9caebcc2cc97847d079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/380d7abacf6e085d82bcf466210906ce8a54fef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/380d7abacf6e085d82bcf466210906ce8a54fef2"}]},"branch":"refs/heads/master"},"0ac2ef7a6ac1e19f55773ee6468372bcb5142756":{"kind":"REWORK","_number":2,"created":"2019-08-01 14:29:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/673484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/84/673484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/84/673484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/84/673484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/84/673484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/84/673484/2"}}},"commit":{"parents":[{"commit":"865a370027ced8240118c61dd387220b22306b03","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/865a370027ced8240118c61dd387220b22306b03"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-30 11:34:49.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-01 14:29:33.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Id9e94865467a9076ee15a9caebcc2cc97847d079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/0ac2ef7a6ac1e19f55773ee6468372bcb5142756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/0ac2ef7a6ac1e19f55773ee6468372bcb5142756"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
