)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~675094","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I230a6b46d91b71881c91b1f4f93d03676a3a0afa","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"static_imports","hashtags":[],"change_id":"I230a6b46d91b71881c91b1f4f93d03676a3a0afa","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-08-07 11:19:34.000000000","updated":"2019-08-08 14:30:34.000000000","submitted":"2019-08-08 14:30:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675094-1565274634318-01c2484d","meta_rev_id":"a9e45fd07ad60dfe85c7680381c0ddd19c512183","_number":675094,"virtual_id_number":675094,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-08 14:30:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-08 12:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-07 21:13:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-08 12:47:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 21:13:41.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-08 12:47:57.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-08 14:30:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f77e39c620473fda83ca1073f0c60ddff44cf81","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-07 11:19:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"770e99dfbe8576082ddefa7cc068840270ae2182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 12:22:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/94/675094/1/check/openstack-ansible-linters/debe244/ : FAILURE in 6m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/94/675094/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3b1ae37/ : FAILURE in 23m 31s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/94/675094/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/e5b6fba/ : FAILURE in 20m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/94/675094/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/bcdf419/ : FAILURE in 33m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/94/675094/1/check/openstack-ansible-deploy-aio_metal-debian-stable/d402aaf/ : FAILURE in 18m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/94/675094/1/check/openstack-ansible-deploy-aio_metal-centos-7/4a702e3/ : FAILURE in 27m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/94/675094/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/b96271b/ : FAILURE in 30m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/94/675094/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f96a511/ : FAILURE in 20m 19s\n- openstack-tox-docs https://logs.opendev.org/94/675094/1/check/openstack-tox-docs/d533b6f/html/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"68c2659597408aead94fde362ca5385160a1ed9b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-07 15:55:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69c39352df16c047aa05413602249585855a8a1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 17:29:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/94/675094/2/check/openstack-ansible-linters/8632169/ : SUCCESS in 7m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/94/675094/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8cb178d/ : SUCCESS in 1h 05m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/94/675094/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/733b3e1/ : SUCCESS in 1h 04m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/94/675094/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/54e48ee/ : SUCCESS in 1h 03m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/94/675094/2/check/openstack-ansible-deploy-aio_metal-debian-stable/663328c/ : SUCCESS in 1h 14m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/94/675094/2/check/openstack-ansible-deploy-aio_metal-centos-7/6c66281/ : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/94/675094/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/9060059/ : SUCCESS in 1h 15m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/94/675094/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/76bc084/ : SUCCESS in 1h 04m 58s\n- openstack-tox-docs https://logs.opendev.org/94/675094/2/check/openstack-tox-docs/60f6141/html/ : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"11e4f239f945d620cb0d7570f7578cdae9c15cdf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-07 21:13:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f013213550d5c0f70b5f730f892599d51857d26c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-08 12:47:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c0338093f93896b3df24c7f033084c6a7a4e8752","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 12:48:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9a7cfbef4bdd817484e7b8f61d65a58d96bd2ce6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:30:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/94/675094/2/gate/openstack-ansible-linters/9556c52/ : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/94/675094/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/4a184fe/ : SUCCESS in 57m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/94/675094/2/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/249a963/ : SUCCESS in 48m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/94/675094/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/52ad58b/ : SUCCESS in 1h 17m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/94/675094/2/gate/openstack-ansible-deploy-aio_metal-centos-7/bd851ca/ : SUCCESS in 1h 38m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/94/675094/2/gate/openstack-ansible-deploy-aio_metal-opensuse-15/f57d3ac/ : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/94/675094/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/64702f4/ : SUCCESS in 1h 19m 58s\n- openstack-tox-docs https://logs.opendev.org/94/675094/2/gate/openstack-tox-docs/e024a62/html/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d5663dc25b53823ba0b03ff7a9afedf2eeccee8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:30:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af1cd0ac8f9f8db7cc84e1c23f26c9b1b7e4bda0","revisions":{"5c8a6e76e022b0e985ff806b8406c27c5c307a31":{"kind":"REWORK","_number":1,"created":"2019-08-07 11:19:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/675094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/94/675094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/94/675094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/94/675094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/94/675094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/94/675094/1"}}},"commit":{"parents":[{"commit":"ee8586edf11b7f60b54b2e1599b81708b032342a","subject":"Merge \"service_setup: refactor service setup to a single file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/ee8586edf11b7f60b54b2e1599b81708b032342a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 11:19:23.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 11:19:23.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like swift-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I230a6b46d91b71881c91b1f4f93d03676a3a0afa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/5c8a6e76e022b0e985ff806b8406c27c5c307a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/5c8a6e76e022b0e985ff806b8406c27c5c307a31"}]},"branch":"refs/heads/master"},"af1cd0ac8f9f8db7cc84e1c23f26c9b1b7e4bda0":{"kind":"REWORK","_number":2,"created":"2019-08-07 15:55:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/94/675094/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/94/675094/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/94/675094/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/94/675094/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/94/675094/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/94/675094/2"}}},"commit":{"parents":[{"commit":"ee8586edf11b7f60b54b2e1599b81708b032342a","subject":"Merge \"service_setup: refactor service setup to a single file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/ee8586edf11b7f60b54b2e1599b81708b032342a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 11:19:23.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 15:55:14.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like swift-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I230a6b46d91b71881c91b1f4f93d03676a3a0afa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/af1cd0ac8f9f8db7cc84e1c23f26c9b1b7e4bda0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/af1cd0ac8f9f8db7cc84e1c23f26c9b1b7e4bda0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
