)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~675380","triplet_id":"openstack%2Fopenstack-ansible-os_designate~stable%2Fstein~I230a6b46d91b71881c91b1f4f93d03676a3a0afa","project":"openstack/openstack-ansible-os_designate","branch":"stable/stein","topic":"static_imports-stable/stein","hashtags":[],"change_id":"I230a6b46d91b71881c91b1f4f93d03676a3a0afa","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-08-08 15:12:17.000000000","updated":"2019-08-12 14:31:43.000000000","submitted":"2019-08-12 14:31:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675380-1565620303186-f62f6123","meta_rev_id":"0da7f644c31233d0ad531a0780285604327575cf","_number":675380,"virtual_id_number":675380,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-12 14:31:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-08-12 12:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-09 21:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-08-12 12:52:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 21:30:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-12 12:52:04.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-12 14:31:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"167b386f3874a0a623697e8e355d4c22adcd5fb0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-08 15:12:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fea7e433c889e8e028434597c6e1a29276e28e80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 16:40:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/80/675380/1/check/openstack-ansible-linters/057937e/ : SUCCESS in 6m 43s\n- openstack-tox-docs https://logs.opendev.org/80/675380/1/check/openstack-tox-docs/f2fb375/html/ : SUCCESS in 2m 56s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 https://logs.opendev.org/80/675380/1/check/openstack-ansible-deploy-aio_metal_designate-centos-7/62ad1a8/ : SUCCESS in 1h 25m 29s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable https://logs.opendev.org/80/675380/1/check/openstack-ansible-deploy-aio_metal_designate-debian-stable/4654406/ : SUCCESS in 1h 06m 10s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic https://logs.opendev.org/80/675380/1/check/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/a195841/ : SUCCESS in 51m 27s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 https://logs.opendev.org/80/675380/1/check/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/df81ec4/ : SUCCESS in 1h 21m 15s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic https://logs.opendev.org/80/675380/1/check/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/69fc617/ : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c1fc8e723089dbe7ab6dc719de524a1881abc323","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-09 21:30:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3826e322404f578e7a5233696c73b4f0bdcbf28e","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-12 12:52:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6756f95075eb3ae5f09ee08478f926ae7f3d50e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 12:52:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65bd200ce1e4c7850b37a1fc934fd836bbf4f078","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:31:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/80/675380/1/gate/openstack-ansible-linters/ad7a796/ : SUCCESS in 4m 03s\n- openstack-tox-docs https://logs.opendev.org/80/675380/1/gate/openstack-tox-docs/4fef4ec/html/ : SUCCESS in 4m 10s\n- openstack-ansible-deploy-aio_metal_designate-centos-7 https://logs.opendev.org/80/675380/1/gate/openstack-ansible-deploy-aio_metal_designate-centos-7/b0a8e18/ : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-aio_metal_designate-debian-stable https://logs.opendev.org/80/675380/1/gate/openstack-ansible-deploy-aio_metal_designate-debian-stable/53ead85/ : SUCCESS in 1h 03m 34s\n- openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic https://logs.opendev.org/80/675380/1/gate/openstack-ansible-deploy-aio_metal_designate-ubuntu-bionic/2a01e39/ : SUCCESS in 56m 17s\n- openstack-ansible-deploy-aio_distro_metal_designate-centos-7 https://logs.opendev.org/80/675380/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-centos-7/e117aa4/ : SUCCESS in 1h 24m 39s\n- openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic https://logs.opendev.org/80/675380/1/gate/openstack-ansible-deploy-aio_distro_metal_designate-ubuntu-bionic/34684ea/ : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":1},{"id":"790e288f7e92733850874fc03f5318a439df8e5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:31:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cff00487f2ece76bcb8ccb36b6645eb76da35df","revisions":{"8cff00487f2ece76bcb8ccb36b6645eb76da35df":{"kind":"REWORK","_number":1,"created":"2019-08-08 15:12:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/80/675380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/80/675380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/80/675380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/80/675380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/80/675380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/80/675380/1"}}},"commit":{"parents":[{"commit":"96cff2e65d3d30b3ec00d81ba39119af3eb384a3","subject":"Merge \"Fix distro installs on Ubuntu\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/96cff2e65d3d30b3ec00d81ba39119af3eb384a3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-07 11:19:23.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-08-08 15:12:06.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like swift-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I230a6b46d91b71881c91b1f4f93d03676a3a0afa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/8cff00487f2ece76bcb8ccb36b6645eb76da35df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/8cff00487f2ece76bcb8ccb36b6645eb76da35df"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
