)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~694303","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I62dcba3a70298060baf975ac012d15e962b92863","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"osa/opendev","hashtags":[],"change_id":"I62dcba3a70298060baf975ac012d15e962b92863","subject":"Replace git.openstack.org with opendev.org","status":"MERGED","created":"2019-11-14 14:14:53.000000000","updated":"2019-11-26 15:39:52.000000000","submitted":"2019-11-26 15:38:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694303-1574782720185-7789a045","meta_rev_id":"77268a6f09eb802eb5b7238bebbf9e0c0071c491","_number":694303,"virtual_id_number":694303,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 15:38:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-11-26 12:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-23 18:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2019-11-26 12:37:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-23 18:27:15.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-26 12:37:58.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-11-26 15:38:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"986a1341b37b431a36bf5df5b8544cf049f33f75","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 14:14:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b5b408a450f81925c7c1dab5e6cbba12b366faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 17:04:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3591993ac43f4bc88f0e1e45d442486f : SUCCESS in 6m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/50b630d7fb5f4dbaba6cc040cca6b3d5 : FAILURE in 58m 47s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dca47561020c40618952502ca8bc3358 : FAILURE in 22m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e8c338020df43cfb688e97202ef2273 : SUCCESS in 1h 00m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8496800cb19b4becadeb4ff836649b16 : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/97e4c80d5dc847ecaf79bf49bb44e420 : SUCCESS in 1h 39m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e8afceb2e2ff4e5f9f8016a219d3e49a : FAILURE in 25m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f14abae850d54d48b7b30bea1f4a659f : SUCCESS in 1h 11m 04s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2bb827f8e30944fbbeeb416fa30eaf50 : SUCCESS in 2h 45m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fce54b89edcf4486a58dc75ed663cab4 : SUCCESS in 2h 04m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db80569260cb4a9a8c99d74a1d6aeb97 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6343b0aabb14bddb524fa3e0526fa3a : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"76a06407af29c3c03d46a08a18b4393e9294a742","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 17:37:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"aec3c9845be8a8d25429cca1beda1afa9469c247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 20:45:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41e699d57e094b469a88c53d2ff0c2e5 : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a3f590a73f20498883dcd0e5c13c3e13 : FAILURE in 59m 17s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0be28246bd6b4bc0b5b8e225ea849db8 : SUCCESS in 1h 01m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c43685755d64b9c8d0af6dcffde50e0 : SUCCESS in 54m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ac1fbf0895b8480d98d7f558940d9871 : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e5657a0969ee460d8e607fc3691b5163 : SUCCESS in 1h 34m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2181aae4a7254173adb3faa8e0254c7c : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f15a88d3ade54384aeaaf240281b63d4 : SUCCESS in 1h 10m 41s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a0132b74935740068fa8019d6dfc2edd : TIMED_OUT in 3h 02m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce4ce530abac45c8859317a0acfbf149 : SUCCESS in 1h 56m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08022c0eb5e44a4cb289f5c5dd2b995e : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528e00e203ee483299c195841a512285 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ca344949cafa98381a879ee5537d7cf9b2b9a4b6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-16 22:09:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dc53242564cdaf4c4cb77c64324c31caed0dac9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-17 00:53:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a29afd75d6e14698a38cfb25a4d42843 : SUCCESS in 4m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/65da2b5692af483fb31f48afc8ee2f16 : FAILURE in 1h 10m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/94faad80d1f04a8eb4d0cd15ec4e2533 : SUCCESS in 1h 00m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a7d6e1c05924e028151aea24a632830 : SUCCESS in 48m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/510999763421481ca8cfd3f82426f2fc : SUCCESS in 1h 18m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/80a2f6adefdc4c629e6be27c87908dea : SUCCESS in 1h 42m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a3641ecaac1742819d07c1eef599853c : SUCCESS in 1h 24m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8a7431df86d4ee386170293a6fef65b : SUCCESS in 1h 09m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c97668252ad4345b5c2006491dd7c10 : SUCCESS in 2h 40m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/268626d8a41a46f988e784e71c13889d : SUCCESS in 1h 50m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776a7e715fba413eaa90b4c68beb9924 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/722e51e32faf45598ef2fd3e1ef1998b : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ab010faa61b1348c2550422552708f3668b2a117","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 09:55:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b49a62ade8cfbf8dd86ce51078943caa27470914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 12:58:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f8e00db17a3d4262aca0041de0a48f57 : SUCCESS in 5m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1a6da070da24907b964aae29318bf34 : FAILURE in 1h 01m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d080d7541b0c4dfb904bc8d5d438f446 : SUCCESS in 1h 03m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2efc21a80e74705ae03c77070122a0e : SUCCESS in 54m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d07dfe912892491b896ef1d8c96e0718 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d5669ea0ae134ea69bf37cc53f553757 : SUCCESS in 1h 47m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d2937455a23349af85cf78fb9c2f6c8c : SUCCESS in 1h 30m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23e7be0c5a8d429290400cd493b3a5ba : SUCCESS in 1h 12m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ea4c344174d4b3c897013831b58c968 : SUCCESS in 2h 56m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3950e3370f0a4ae19137ed04158d8b65 : SUCCESS in 1h 59m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567c000322854209adf54528fab70519 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23ac31f7dae749349201fe54b0d6325e : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"38223e445709e701c8dc25910e032e241f6abdbb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-20 09:29:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31f1ba310caef288c6a0a7396470c5dbf3702d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 12:44:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1dfb623755ec4763b5598512ce4aadb1 : SUCCESS in 8m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7e684c25118e46a7b85921b6d6cf634a : FAILURE in 47m 47s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/518a094cb48b483f90445de194314b8a : SUCCESS in 1h 00m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa712b328b4b4eb9bf9976f41a16bc8d : SUCCESS in 52m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d5df181a51c8425ea12c420b1dcb9667 : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b049345b29434ddea81f20148abf313e : SUCCESS in 1h 39m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5894c6f0017b4b2eb120a6a598959916 : SUCCESS in 1h 25m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83e9753ee117460784043244d96543d7 : SUCCESS in 1h 22m 34s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fdd7c73231eb4a6d920762fd91051123 : TIMED_OUT in 3h 02m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20954a3f09bb4a3e9ce025576315053d : SUCCESS in 1h 51m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fafecddb72f458e9596b06ef7e679dc : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b434255c5af4c83b149e0b2f29e42e6 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"21fee37d96ba5a57d926cf10362cd5928ed98455","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-22 20:44:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3d935439d30ddf7bccb8cbd3faabf0af1d7b9152","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 23:20:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/493aeae9f91b4cadbdc9d1af12d96a00 : SUCCESS in 5m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9ffa8d8aff3844f7aeae79461e71f238 : SUCCESS in 1h 18m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/12ad97abc6c54dd8a32c62a6a7577597 : SUCCESS in 1h 02m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7ce8fee5ed74056854b1c7d52377526 : SUCCESS in 49m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a9e12ec8ea2941c5916c284fc57f543a : SUCCESS in 1h 14m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c7dfbd98572b4642acfd7af88c9fc246 : SUCCESS in 1h 45m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fef578b0afac412181c7119c1fc74a17 : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aff130a2c5c54888b2ce05c007b0d9b1 : SUCCESS in 1h 12m 09s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4b4544e7823c421cb4b22e49c301d927 : SUCCESS in 2h 34m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab8811442290417199abf45d54fddf07 : SUCCESS in 1h 49m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44408ab92fb4033aaed9b585e2d347f : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7daabc3dc3447ceafbc49a98439f5d2 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"36dff6e1d68377d0a51b872fe30306d2d2837516","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-23 18:27:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0cb1f6c1c16863a3bc159fbcdae4d13cac254825","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-11-26 12:37:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6c52cdf68f44183796afd24d9db11d37f68b35fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 12:38:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e7fc11a48edc5e44bd24f70147d7a8d3d19e6996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:38:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b7c3ef8f04de0b1619573067e739d265277e6ab7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:38:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5812f8156fde43e192995b83f6a34636 : SUCCESS in 7m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0795d40cc84b4a129363df26d85c2999 : SUCCESS in 1h 18m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69abac9e36314ebe90964e1e46f2789f : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7136fa76db2240f59ff15fcc3de4f48a : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d44a105cb614a098e08227d16e5219a : SUCCESS in 1h 43m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5a754219f7694209aa9c7c0ea85e6432 : SUCCESS in 1h 39m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/844d2b110cad4b0c93fe3005ada3980a : SUCCESS in 1h 28m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f8cce96d2fd04786aaeb6a6235b9f34c : SUCCESS in 2h 55m 25s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5140395d5b1440138ed6176e717a494e : SUCCESS in 2h 18m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b1173aa0534ff7a8813a8dd98ca768 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bea347b86a2459c873947e94ab49f5b : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"1b5a3ca8e2f5879dd83e83c389418290db08b9cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:39:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb04d0ad94eb4b25a44e4c088e887fe6 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93ab9cb03454edcb952054b4b2625ea : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf7fe362e656038514932b401ef28d887e4fafd3","revisions":{"f1d39d18385a36b9268502c2979b504801496366":{"kind":"REWORK","_number":1,"created":"2019-11-14 14:14:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/694303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/03/694303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/1"}}},"commit":{"parents":[{"commit":"c76178db846b452637dbb9daa3b28364405c9e48","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c76178db846b452637dbb9daa3b28364405c9e48"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 14:14:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 14:14:32.000000000","tz":120},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\nAlso we change upper constraints url due to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I62dcba3a70298060baf975ac012d15e962b92863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/f1d39d18385a36b9268502c2979b504801496366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/f1d39d18385a36b9268502c2979b504801496366"}]},"branch":"refs/heads/master"},"eda141a30edef258ab35658a97e89ba14dbf2559":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-14 17:37:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/694303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/03/694303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/2"}}},"commit":{"parents":[{"commit":"c76178db846b452637dbb9daa3b28364405c9e48","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c76178db846b452637dbb9daa3b28364405c9e48"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 14:14:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-11-14 17:37:47.000000000","tz":0},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\nAlso we change upper constraints url due to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nDepends-On: https://review.opendev.org/693841/\nChange-Id: I62dcba3a70298060baf975ac012d15e962b92863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/eda141a30edef258ab35658a97e89ba14dbf2559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/eda141a30edef258ab35658a97e89ba14dbf2559"}]},"branch":"refs/heads/master"},"bf7fe362e656038514932b401ef28d887e4fafd3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-22 20:44:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/694303/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/03/694303/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/03/694303/3"}}},"commit":{"parents":[{"commit":"41bf7d522c9d5d596bf00e73f858ede0c56157b0","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/41bf7d522c9d5d596bf00e73f858ede0c56157b0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 14:14:32.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-11-22 20:44:36.000000000","tz":0},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\nAlso we change upper constraints url due to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nDepends-On: https://review.opendev.org/693841/\nChange-Id: I62dcba3a70298060baf975ac012d15e962b92863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/bf7fe362e656038514932b401ef28d887e4fafd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/bf7fe362e656038514932b401ef28d887e4fafd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
