)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~713053","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I152220d4c440202de1a61b1aee891bdb659e5577","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"memcached-refactor","hashtags":[],"change_id":"I152220d4c440202de1a61b1aee891bdb659e5577","subject":"Refactor memcached_servers","status":"MERGED","created":"2020-03-13 19:37:32.000000000","updated":"2020-03-16 16:05:11.000000000","submitted":"2020-03-16 16:02:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713053-1584374529262-4545ba28","meta_rev_id":"9a56470e40fadc5fa70c4076a092e5b4a51b05be","_number":713053,"virtual_id_number":713053,"owner":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-16 16:02:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-03-16 14:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-14 16:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-03-16 14:12:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-14 16:14:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-16 14:12:03.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-03-16 16:02:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4638fe3ff80879642f6d4d24d62ce2922447ee94","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-03-13 19:37:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dd4dd7b20a2a6ff57a1311b155153d72c14480f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:50:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/91aed845108e4de1bad943b2a090ba3f : FAILURE in 9m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b7d62465ef38474193ea69f54ea578b0 : FAILURE in 24m 57s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6c384b956aa248fdb16ec2d9028baf67 : FAILURE in 27m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3bea76088ab480e9cb153bb72487282 : FAILURE in 25m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/eb1d6900e4e54f2e8b725f7ded57dbc4 : FAILURE in 24m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae1078952734437ab6c3162ff5fd8a59 : FAILURE in 30m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/56962e8dcb2a485ab2df840d63aeda1c : FAILURE in 31m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a07deed00a14c3f96011fa4b7cd75db : FAILURE in 26m 03s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1551b891ba64beb85e08015f30d2fd9 : FAILURE in 9m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/700015c8715c45f89c225aac3bc94706 : SUCCESS in 2h 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c022f3bfaa4d76b7a32f13dea9dc1d : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7f583540ea0f2818592cf6057d77f4db931ebff5","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-03-13 22:21:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53f3cfd3ce06f2bcb1c716e7aa50a24e76b5804f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 00:20:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/56ce405ff0f644ac9e9edff58e0ccee9 : SUCCESS in 7m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b0fecdcea3274867a4ace43016639c27 : FAILURE in 52m 07s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6251b4bb943145e681c6b995b8fd1622 : FAILURE in 30m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dbe1d61c4a94e0b9dfb1191d5bf9016 : FAILURE in 43m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1f18d8eb3b6b4b1a9faad44c4f8bf41c : FAILURE in 55m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d34cd44e24e493f98a5be01bd725a37 : FAILURE in 1h 23m 27s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f42513af022a47d6a03cf7ece09578ef : FAILURE in 1h 09m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/871b557e936847d6bb50c500c2ee36bb : FAILURE in 1h 05m 30s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1bb0f6f5a2184753be8e04cf166b315a : FAILURE in 9m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54c6202c2c184527a5853cb77353d433 : SUCCESS in 1h 55m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69392a8260d64a4593f94f8ad9a696b1 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3ddc5299d351c20f5d8491527f5b8dadec9baa36","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-14 12:03:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"913602ab5cbb77faa16fa8016ba2c3375ef34b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 14:04:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7364e4637b864d77a2cfd425220969a3 : SUCCESS in 5m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6d0009b92de9401e9d31eed0fe2edd6f : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c20daebad759426c893a2b086dcd80de : FAILURE in 29m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c43e8e17428741b395a048efcf9fe5f5 : SUCCESS in 1h 13m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b3867db5f8214aa69df0a657e40d8211 : SUCCESS in 1h 06m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dce737135bcd4bb29f48a01a7a9251cf : SUCCESS in 1h 47m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c657d4e3aa914ad4a61b94423c862106 : SUCCESS in 1h 17m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2545ce522ef4829b27aa285414ae39b : SUCCESS in 1h 09m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/38a89d6f9f6d46eebe2313551631fa10 : FAILURE in 7m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24306e00a6244cd198be182ff12097da : SUCCESS in 1h 59m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fad7cac52fa43c3b5d77a08bd146b97 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"65258158f5f9628f49ec6b2ce936fef61631e31f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-14 16:14:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ec7c8d2a4b5c37295061b2f53da270f9bd2dcce8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-16 14:12:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"19c868623fc9c367b2f1ced8d46b24bbc446ce87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 14:12:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"59ba91c59d788ecf8695c9b0074ce99d8eb6bf4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 16:02:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a1f9f771e8aa4c71cb8c2181532ca7ebc6e06d91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 16:02:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bd29a4a3423e416b839beb7fb8f97881 : SUCCESS in 9m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de1e3f26e58d499c92ad6d2f6ef70e8c : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/977b7e465d9647ef9104c816d4e56e31 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/03ef6d87fd0045e2bfc364fce9a92d4f : SUCCESS in 1h 29m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4a445d8e812e461fadb6eec3fb3bec2d : SUCCESS in 1h 47m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b22acc29695d4d989029d4534331d893 : SUCCESS in 1h 28m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d2e2a47b45d4faaaf564ad779a36c9d : SUCCESS in 1h 34m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2db878d05c4ba1ad83a0d1b0748f42 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":3},{"id":"fa6cdef3a78c1e5908085aa37dd10384a5fae18f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 16:05:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c95e2758d34ffbaee9712464e2fd78 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7f9c47c209773c4f27fd1b84f15d8a71aafcb2c0","revisions":{"cc66346a559b89accbead4a56db615c628f2574b":{"kind":"REWORK","_number":1,"created":"2020-03-13 19:37:32.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/53/713053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/53/713053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/1"}}},"commit":{"parents":[{"commit":"6b53c22fa6d1795ff4d12abd22948e3e700d7f26","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/6b53c22fa6d1795ff4d12abd22948e3e700d7f26"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-13 19:37:06.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-13 19:37:06.000000000","tz":0},"subject":"Refactor memcached_servers","message":"Refactor memcached_servers\n\nThis patch aims to add a prefix for memcached_server\non each role to give the ability for deployers to\noverride the location of memcached cluster. I.e users\nwants to create a single memcached cluster with k8s\nfor each service.\n\nChange-Id: I152220d4c440202de1a61b1aee891bdb659e5577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/cc66346a559b89accbead4a56db615c628f2574b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/cc66346a559b89accbead4a56db615c628f2574b"}]},"branch":"refs/heads/master"},"d7076ee9aed074cebeb8846fb820bbb0ec095f3c":{"kind":"REWORK","_number":2,"created":"2020-03-13 22:21:12.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/53/713053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/53/713053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/2"}}},"commit":{"parents":[{"commit":"6b53c22fa6d1795ff4d12abd22948e3e700d7f26","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/6b53c22fa6d1795ff4d12abd22948e3e700d7f26"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-13 19:37:06.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-13 22:20:58.000000000","tz":0},"subject":"Refactor memcached_servers","message":"Refactor memcached_servers\n\nThis patch aims to add a prefix for memcached_server\non each role to give the ability for deployers to\noverride the location of memcached cluster. I.e users\nwants to create a single memcached cluster with k8s\nfor each service.\n\nWe also add pymemcache based on [1]\n\n[1] https://review.opendev.org/711429\n\nChange-Id: I152220d4c440202de1a61b1aee891bdb659e5577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/d7076ee9aed074cebeb8846fb820bbb0ec095f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/d7076ee9aed074cebeb8846fb820bbb0ec095f3c"}]},"branch":"refs/heads/master"},"7f9c47c209773c4f27fd1b84f15d8a71aafcb2c0":{"kind":"REWORK","_number":3,"created":"2020-03-14 12:03:05.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/53/713053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/53/713053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/53/713053/3"}}},"commit":{"parents":[{"commit":"6b53c22fa6d1795ff4d12abd22948e3e700d7f26","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/6b53c22fa6d1795ff4d12abd22948e3e700d7f26"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2020-03-13 19:37:06.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-14 12:03:03.000000000","tz":0},"subject":"Refactor memcached_servers","message":"Refactor memcached_servers\n\nThis patch aims to add a prefix for memcached_server\non each role to give the ability for deployers to\noverride the location of memcached cluster. I.e users\nwants to create a single memcached cluster with k8s\nfor each service.\n\nWe also add pymemcache based on [1]\n\n[1] https://review.opendev.org/711429\n\nChange-Id: I152220d4c440202de1a61b1aee891bdb659e5577\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/7f9c47c209773c4f27fd1b84f15d8a71aafcb2c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/7f9c47c209773c4f27fd1b84f15d8a71aafcb2c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
