)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~719746","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I272ca3391f3ddae2306d8c8513c624e976b26191","project":"openstack/openstack-ansible-os_designate","branch":"master","hashtags":[],"change_id":"I272ca3391f3ddae2306d8c8513c624e976b26191","subject":"Cleanup py27 support","status":"MERGED","created":"2020-04-14 07:21:09.000000000","updated":"2020-04-14 19:24:59.000000000","submitted":"2020-04-14 19:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719746-1586892152893-743c3807","meta_rev_id":"e0625faec1e461c1a955872f93ca9f4b7fcb3ffd","_number":719746,"virtual_id_number":719746,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 19:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2020-04-14 08:47:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-14 09:09:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-04-14 19:11:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-14 09:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-04-14 19:11:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 08:47:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-14 09:09:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-14 09:15:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-14 19:11:05.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-04-14 19:22:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2d5530cba9175f63f8c69ed2ec3d2316d049d67","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-04-14 07:21:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1942391980c31c786bab782e0a6dc1b80e3432f7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-14 07:29:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67482a2d70e8a65969d547215619046421e43416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 07:43:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1637b3b0cbb44f06b9a469e511c4ef1f : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddbbe216bcc24f16b428007ab0835d58 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eba3752170de9bf212eca86acba67e8c068c90bb","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-04-14 08:32:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3da8d46a933a8f68947f2c397ed2582afa85585a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 08:43:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4e873b367a1b4878aa6760fff4faaea0 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4436480630734142a57c9ba0c9d27778 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca7f5122118a4efdb72dfbef903a53f7 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f3befe807795aaf2f5ddad944a6cbbf7797b03c0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-14 08:47:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"665bcfbde471feb88631ecf3a543f56aca5aeefa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-14 09:09:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d173fc2942b5beaa23ce38d050a1716331ce8bae","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-14 09:15:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1df10fe0fcd1e9053a4ca69bdb1008869b9da051","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-04-14 19:11:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2d20bcd1fe2f08d477fe9e665b4617f0328894ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 19:11:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4aa4307606a2960478b75f11b9b358379fe97803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 19:22:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f116a476242641b694ec82c874cfe4c9 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2e3f331adc45f59a63d929b113fb31 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8fb3e690b824057839d8447ce02fa49 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"cafdcb208674652ed304d651f6048486ceee5f0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 19:22:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e0625faec1e461c1a955872f93ca9f4b7fcb3ffd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 19:24:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5399c33b2246cea5da3f1505d81c26 : SUCCESS in 2m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c829eb30182b4f309221fae8d62a4333 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"354a8a5ba08a370a8f09227fbbeb9749bf294390","revisions":{"c0c8b056fa24d716926d8b4ab2acf905d15c852b":{"kind":"REWORK","_number":1,"created":"2020-04-14 07:21:09.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/46/719746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/46/719746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/46/719746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/46/719746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/46/719746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/46/719746/1"}}},"commit":{"parents":[{"commit":"fb77405055fa5421d12d933bda00f7a734d1f91a","subject":"Merge \"Fix paging link protocol when behind haproxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/fb77405055fa5421d12d933bda00f7a734d1f91a"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-04-14 07:20:16.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-04-14 07:20:47.000000000","tz":480},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Cleanup doc/source/conf.py to remove now obsolete content.\n\nChange-Id: I272ca3391f3ddae2306d8c8513c624e976b26191\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c0c8b056fa24d716926d8b4ab2acf905d15c852b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/c0c8b056fa24d716926d8b4ab2acf905d15c852b"}]},"branch":"refs/heads/master"},"354a8a5ba08a370a8f09227fbbeb9749bf294390":{"kind":"REWORK","_number":2,"created":"2020-04-14 08:32:13.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/46/719746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/46/719746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/46/719746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/46/719746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/46/719746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/46/719746/2"}}},"commit":{"parents":[{"commit":"fb77405055fa5421d12d933bda00f7a734d1f91a","subject":"Merge \"Fix paging link protocol when behind haproxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/fb77405055fa5421d12d933bda00f7a734d1f91a"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-04-14 07:20:16.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-04-14 08:31:42.000000000","tz":480},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Cleanup doc/source/conf.py to remove now obsolete content.\n\nChange-Id: I272ca3391f3ddae2306d8c8513c624e976b26191\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/354a8a5ba08a370a8f09227fbbeb9749bf294390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/354a8a5ba08a370a8f09227fbbeb9749bf294390"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
