)]}'
{"id":"openstack%2Fopenstack-ansible-os_designate~734193","triplet_id":"openstack%2Fopenstack-ansible-os_designate~master~I98cf44744f46bea1392d8f35b19c97c3be51bbfe","project":"openstack/openstack-ansible-os_designate","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I98cf44744f46bea1392d8f35b19c97c3be51bbfe","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-06-08 19:54:34.000000000","updated":"2020-06-17 23:58:22.000000000","submitted":"2020-06-17 23:56:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734193-1592438166175-4bbebc21","meta_rev_id":"74828dafeeb6c2dcc5cc633dd96093a535858fc5","_number":734193,"virtual_id_number":734193,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 23:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-09 05:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-09 05:55:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 05:55:41.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-17 23:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc3184a3d64d6355c0d655b7111ec99db624f61b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-08 19:54:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0713f9ac109a411e48c3bc52a0a9d7eec5d34e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 22:55:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4eb28d0e69e04e15883ef80dea8ab071 : SUCCESS in 11m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e1ff43dd3539409b96c6af40a34b5a44 : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2802966b6f61407390333bba62978f07 : SUCCESS in 1h 03m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c453b5d6d2714c1eb3f28076f3a75ede : SUCCESS in 1h 12m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e71dcd58d5af4b8f990610651be7d975 : SUCCESS in 1h 37m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/126b9fe6156c45b7be76840d68a11306 : SUCCESS in 1h 25m 51s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e85adc3c1a56496ab8d9c3fde5385bf8 : SUCCESS in 2h 44m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cea786f81f54ab3bcf405cf5878c4e3 : SUCCESS in 2h 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135305ae8179405080a67d1e3fa407a9 : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cbe4c4d7999d241c0faaf54ed39b87ddb65a86f3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-09 05:52:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"478338304352d13f869469711356e429ca5d8d07","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-09 05:55:41.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06f5e518f764319d4cd30adb3532b001f06cc59e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 05:55:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df8a37bd6ebb81ddab8345ad2369f110f36a17f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 07:43:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/880aee28e0a34a7380302a93123e70f5 : SUCCESS in 9m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f8dca231674b4fefa9afbb84c4667052 : SUCCESS in 1h 28m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/364057ebc03b4a25ac2eda46355205ec : SUCCESS in 1h 11m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/be1ff2a09f96410db0e83bbfe0e4dc23 : SUCCESS in 1h 16m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e6656703c586461abce420825f898195 : SUCCESS in 1h 41m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2837f668c6244d33a7d7be99a8282b06 : FAILURE in 58m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4feb2e1d65417d9b1af45f7f0bde00 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1422494288e79f7258a8f4131b22215ad3ca75b0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 19:52:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"baab008687a10980fa0151076dc3d2201f451421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:14:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3996c829ae4247db9253afc4affbcc82 : SUCCESS in 9m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cc670a77ce4048b1905adbdc0f3f19f4 : SUCCESS in 1h 13m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c08d3b85e714a50822439e81c27c314 : SUCCESS in 1h 02m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ace7f9159f534b63a552d638ee5e4366 : SUCCESS in 1h 17m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/281bc333a8574f098a64e4b89683f5da : SUCCESS in 1h 34m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2231164a1d147ecb702fdd863374f74 : SUCCESS in 1h 20m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c6fccd0f15e54c6da0530c2ae81041b2 : FAILURE in 1h 50m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2027dbe7ba3141a0b333e38a25a97156 : SUCCESS in 2h 03m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63029587c95f404bac19ae774cbc2dcc : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":1},{"id":"bee84c481d1fcff6792f37fb52dfc712f98d3570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:15:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b602b733a5900f4ddd7a8909d1c36c7f779d44b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:56:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c4dbe7eb506c4ff586c27c0a45e915ce : SUCCESS in 11m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5ce75df4213f45ab8303508f6dc5b951 : SUCCESS in 1h 07m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a92633fa46d49358f2489fdab1a0a0d : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/44e377991dc24ef1b6a138bb0e8c4130 : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df18fdd8608141ee816cb86bf32a3b82 : SUCCESS in 1h 33m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec454a7f5cec42deb39083454a1fd548 : SUCCESS in 1h 14m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fadd101e641b4f04982a9e822b15131f : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e8281bbd5a9e2303cfe23363766f41bf68a75da7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:56:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3da3fdcb84c037a541e2fd0e91d74fd2d6cac33d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:58:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a8cdeea79b4ee788c2cfb9f782a4b3 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4eeb53bad99dfa008e604a08cf6caa8124883cce","revisions":{"4eeb53bad99dfa008e604a08cf6caa8124883cce":{"kind":"REWORK","_number":1,"created":"2020-06-08 19:54:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/734193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_designate","ref":"refs/changes/93/734193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/93/734193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/93/734193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/93/734193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_designate refs/changes/93/734193/1"}}},"commit":{"parents":[{"commit":"539d67fe911d98615c660c5e12ce68f203dee97d","subject":"Merge \"Use newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/539d67fe911d98615c660c5e12ce68f203dee97d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-08 19:54:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-08 19:54:32.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I98cf44744f46bea1392d8f35b19c97c3be51bbfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/4eeb53bad99dfa008e604a08cf6caa8124883cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_designate/commit/4eeb53bad99dfa008e604a08cf6caa8124883cce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
